)]}'
{"id":"openstack%2Felection~941938","triplet_id":"openstack%2Felection~master~Iecbbb26e566313f1f27dddfde6e27c96d60251ce","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-02-18 09:16:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iecbbb26e566313f1f27dddfde6e27c96d60251ce","subject":"Add Artem Goncharov candidacy for OpenStackSDK PTL","status":"MERGED","created":"2025-02-17 15:50:07.000000000","updated":"2025-02-18 09:18:58.000000000","submitted":"2025-02-18 09:16:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941938","meta_rev_id":"94348fd6227a528700c284eef764ac3bd5edcae5","_number":941938,"virtual_id_number":941938,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-18 09:16:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2025-02-17 21:55:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2025-02-18 08:24:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2025-02-18 08:24:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-17 16:11:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-17 21:55:22.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2025-02-18 08:24:58.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"4b9111b1184e25a41afb782dcf48eec1dbacc0b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-02-17 15:50:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc21ff9725ad4c6b1ecd81fd6ec3b8a524c48122","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-02-17 15:54:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"20520ae552c7f87605651cd523f8b5326bde082a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 16:11:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9fcdcdf4e8c49ca85fef96d3cd8eb50\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7819d4174af9464592d80a3e2d1c0c7b : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fcee233b8d34f13a31f1df39cdebe25 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b24b0d7f4ec4ed5b414cbdd3f7879b1 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01c01c91dc8a4c3fa6c5c6f63da8e59c : SUCCESS in 3m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/eba32773a92f494f921af87c71440c09 : SUCCESS in 2m 57s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/8c4d919c26214a3386c562c026182ffc : SUCCESS in 3m 12s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/10ddbc7b21264236ac4ae4a30daf5c5a : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"d167118f65449c5d8ba9fca65235a7523d65e9a4","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-02-17 21:55:22.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"68242c867cdbb095d96b43fbe7c41f8f836cdada","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-02-18 08:24:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b6271b131b293e30dba29314cd552bd68c3355f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 08:25:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b6d3ea02246e7c66b55fb48bd259441b85058a36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 09:16:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5e187502d434d5a999943880bfb4429\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae68def59d3848a98b7086f6713f9e61 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff45e8a5cecd41b9a8d9f53ace593a1f : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5853cbc84c44a80b7fb490eb6f10e37 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f5eb8ebf59634d90a4b7187c5b6904b6 : SUCCESS in 4m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6c960d1dbcb14ba4b5e18f73ddce25db : SUCCESS in 3m 32s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/f8c0c675726a4198bc4b07c2896537e9 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":2},{"id":"dc7b3b2f86ce5d77137f36730811cd8f3e2babe2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 09:16:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"94348fd6227a528700c284eef764ac3bd5edcae5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 09:18:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e2d64a033e74623a2bbf24a2fb24f3f\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/2e7ddbbf8bba445ead7e14582c8ab0aa : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6951ebfd63e06e5c7cdb630b88607f5f82ed7e29","revisions":{"46fffb56685fcb26c85ed3e7cdb71fa6cd8ac2b5":{"kind":"REWORK","_number":1,"created":"2025-02-17 15:50:07.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/38/941938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/38/941938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/38/941938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/38/941938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/38/941938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/38/941938/1"}}},"commit":{"parents":[{"commit":"305248d35aef892ac827b83ad03d3ad11dafa5e6","subject":"Add Brian Haley (haleyb) candidacy for Neutron PTL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/305248d35aef892ac827b83ad03d3ad11dafa5e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-02-17 15:49:21.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-02-17 15:49:21.000000000","tz":60},"subject":"Add Artem Goncharov candidacy for OpenStackSDK PTL","message":"Add Artem Goncharov candidacy for OpenStackSDK PTL\n\nChange-Id: Iecbbb26e566313f1f27dddfde6e27c96d60251ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/46fffb56685fcb26c85ed3e7cdb71fa6cd8ac2b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/46fffb56685fcb26c85ed3e7cdb71fa6cd8ac2b5"}]},"branch":"refs/heads/master"},"6951ebfd63e06e5c7cdb630b88607f5f82ed7e29":{"kind":"REWORK","_number":2,"created":"2025-02-17 15:54:31.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/38/941938/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/38/941938/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/38/941938/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/38/941938/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/38/941938/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/38/941938/2"}}},"commit":{"parents":[{"commit":"305248d35aef892ac827b83ad03d3ad11dafa5e6","subject":"Add Brian Haley (haleyb) candidacy for Neutron PTL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/305248d35aef892ac827b83ad03d3ad11dafa5e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-02-17 15:49:21.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-02-17 15:54:23.000000000","tz":60},"subject":"Add Artem Goncharov candidacy for OpenStackSDK PTL","message":"Add Artem Goncharov candidacy for OpenStackSDK PTL\n\nChange-Id: Iecbbb26e566313f1f27dddfde6e27c96d60251ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/6951ebfd63e06e5c7cdb630b88607f5f82ed7e29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/6951ebfd63e06e5c7cdb630b88607f5f82ed7e29"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
