)]}'
{"id":"openstack%2Felection~941942","triplet_id":"openstack%2Felection~master~Ic3d5ef7cdc577f4fd5c2f7fa0773f7d8243d1aeb","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"5241":{"account":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"last_update":"2025-02-24 15:50:18.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic3d5ef7cdc577f4fd5c2f7fa0773f7d8243d1aeb","subject":"Add Martin Magr candidacy for PTL of Telemetry project","status":"ABANDONED","created":"2025-02-17 16:42:14.000000000","updated":"2025-02-24 15:50:18.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ce0eb11040de90482a952d35508aa59ccf1e0421","_number":941942,"virtual_id_number":941942,"owner":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-17 18:40:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":-1,"date":"2025-02-17 22:00:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-17 17:31:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-02-17 18:40:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-17 22:00:27.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2025-02-18 08:58:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REMOVED"}],"messages":[{"id":"01aca6d561f8543416b71dfc7234857f8e2104b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2025-02-17 16:42:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"219d81b6074d719f5a48bc5ce31daf5359cbd87a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-17 17:31:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c686f5ac3c4702b3e6e5ceab2ccd5a9de661847b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 18:40:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/df3f0da5adde4c049a1a5de245664470\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2876422f7b9b4ae48006e8f027fff6a5 : FAILURE in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/395c31654f294fb3adcc1b2a2faf4d51 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e611d2884524f1fbe4e91d01b4c6236 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44896883ddd1426ab74f20e540731513 : SUCCESS in 2m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5a96256d2cae4ff0b2f0f6668d3db35c : SUCCESS in 2m 17s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/fbf42b0d22fe4963be3131fe9ed81aed : FAILURE in 2m 17s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/3df0352576ed42cda75373c2586aed88 : FAILURE in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"0e5ae00de6aecf9bb84a265b31a27106a1dd2e82","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-02-17 22:00:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41d1ed6d025e9e46ac03d419135b65a490583068","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-18 08:58:09.000000000","message":"Removed Code-Review+1 by \u003cGERRIT_ACCOUNT_37598\u003e\n","accounts_in_message":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"_revision_number":1},{"id":"ae9fadb6823b96b7e250b60d548858d6b5a5b615","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-18 08:58:19.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_37598\u003e.","accounts_in_message":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"_revision_number":1},{"id":"ce0eb11040de90482a952d35508aa59ccf1e0421","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2025-02-24 15:50:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69a8e0d614b16ebc04695408e8d60d8363fcf9f1","revisions":{"69a8e0d614b16ebc04695408e8d60d8363fcf9f1":{"kind":"REWORK","_number":1,"created":"2025-02-17 16:42:14.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/42/941942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/42/941942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/42/941942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/42/941942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/42/941942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/42/941942/1"}}},"commit":{"parents":[{"commit":"305248d35aef892ac827b83ad03d3ad11dafa5e6","subject":"Add Brian Haley (haleyb) candidacy for Neutron PTL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/305248d35aef892ac827b83ad03d3ad11dafa5e6"}]}],"author":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2025-02-17 16:40:48.000000000","tz":60},"committer":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2025-02-17 16:42:12.000000000","tz":60},"subject":"Add Martin Magr candidacy for PTL of Telemetry project","message":"Add Martin Magr candidacy for PTL of Telemetry project\n\nChange-Id: Ic3d5ef7cdc577f4fd5c2f7fa0773f7d8243d1aeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/69a8e0d614b16ebc04695408e8d60d8363fcf9f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/69a8e0d614b16ebc04695408e8d60d8363fcf9f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
