)]}'
{"id":"openstack%2Felection~942075","triplet_id":"openstack%2Felection~master~I6a6658ceed229cd28ce42efc161b6c1f2b2e3e43","project":"openstack/election","branch":"master","topic":"ptl-candidacy","attention_set":{},"removed_from_attention_set":{"10366":{"account":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"last_update":"2025-02-18 13:52:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6a6658ceed229cd28ce42efc161b6c1f2b2e3e43","subject":"Adding Hemanth Nakkina PTL candidacy for Sunbeam","status":"MERGED","created":"2025-02-18 10:04:40.000000000","updated":"2025-02-18 21:55:59.000000000","submitted":"2025-02-18 13:52:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942075-ptl-candidacy","meta_rev_id":"6180ce21416435d2fd09f28c48908f559eaeba4e","_number":942075,"virtual_id_number":942075,"owner":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-18 13:52:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"recommended":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"all":[{"value":1,"date":"2025-02-18 21:55:59.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},{"value":2,"date":"2025-02-18 10:48:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2025-02-18 13:38:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2025-02-18 13:38:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-18 10:26:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-18 10:48:30.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2025-02-18 13:38:08.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-02-18 21:55:59.000000000","updated_by":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"reviewer":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"state":"REVIEWER"}],"messages":[{"id":"ac636c6a94a531fb3bd530cf58c223d324c7dfa8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2025-02-18 10:04:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"342b529f6ff6cb49c87aa4ef1b45e0931e7083d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 10:26:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdd294af703448a595b47aad0cae1ba5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91134d6870994e51a4384382fa590a8a : SUCCESS in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce16091bcdb74cd8aceeadc74edff870 : SUCCESS in 2m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78c15fec58da4268b57266dd946074cf : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7cab3196cec14596982f03ba18f571e6 : SUCCESS in 2m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f175a2b13e0148deb2d644420c46249b : SUCCESS in 3m 11s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/3b7a918b69c14536a8b8a35e21aae23f : SUCCESS in 5m 42s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/227ad806a76f462497e6889965df1c6f : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"c61ed0282b81b54c9ae19d43789c84987ed7b5bc","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-02-18 10:48:30.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f5c09715e14d1fab5c99298a6287218aa8563c7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-02-18 13:38:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"85ea8e8e739ddf96184ae0ba5b345c9c995c67ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 13:38:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac6eb5e8dbc500b4f293c60a52e2fd5d1550f564","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 13:52:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74708956bf884ee0ae6c5cee65aeeaf4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2979b31643b64cfe9ae2e4cbd0bc1642 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0d10b774e804b9db4549ba5f7f99ff0 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed02bb7e6b6449f7906d9569a0b42789 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/917cbd597f6a4485b46f2930f5b8bddb : SUCCESS in 3m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7eebe36b05214e22a9efe4dd614c885f : SUCCESS in 3m 19s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/5e9ba1a7f0754afb80b85b1fd1dd6ac5 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1},{"id":"da06670694b29bff50c07f9e9d5f40ca86ec9dfb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 13:52:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5a1aaf8929cb1efb10871b10484964b362fed29a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 13:53:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1cc7f2a05fe4efc95b4005bedb603bf\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/b41346c338234decba8e0982d7c40a56 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1},{"id":"6180ce21416435d2fd09f28c48908f559eaeba4e","author":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"date":"2025-02-18 21:55:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9d3db1b70d97930c159f746f984d979c54c6d18c","revisions":{"9d3db1b70d97930c159f746f984d979c54c6d18c":{"kind":"REWORK","_number":1,"created":"2025-02-18 10:04:40.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/75/942075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/75/942075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/75/942075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/75/942075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/75/942075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/75/942075/1"}}},"commit":{"parents":[{"commit":"5d92f9a5864fe5d091d04418cfc3fafa5ac3944b","subject":"Merge \"Add Artem Goncharov candidacy for OpenStackSDK PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/5d92f9a5864fe5d091d04418cfc3fafa5ac3944b"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2025-02-18 10:02:41.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2025-02-18 10:04:36.000000000","tz":330},"subject":"Adding Hemanth Nakkina PTL candidacy for Sunbeam","message":"Adding Hemanth Nakkina PTL candidacy for Sunbeam\n\nChange-Id: I6a6658ceed229cd28ce42efc161b6c1f2b2e3e43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/9d3db1b70d97930c159f746f984d979c54c6d18c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/9d3db1b70d97930c159f746f984d979c54c6d18c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
