)]}'
{"id":"openstack%2Felection~942234","triplet_id":"openstack%2Felection~master~I72f011dc897cd9bd92885961238b043559d431bc","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-02-19 22:09:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I72f011dc897cd9bd92885961238b043559d431bc","subject":"Add Artem Goncharov candidacy for TC","status":"MERGED","created":"2025-02-19 18:58:53.000000000","updated":"2025-02-19 22:10:38.000000000","submitted":"2025-02-19 22:09:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942234","meta_rev_id":"c1929fdc44e237401ea8d3d81cc0507360e29c16","_number":942234,"virtual_id_number":942234,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-19 22:09:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2025-02-19 21:45:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2025-02-19 21:14:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2025-02-19 21:59:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-19 19:09:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-19 21:14:06.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-02-19 21:45:10.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"55aa6ca57744d02d84c0b4d01076e60d46db3ead","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-02-19 18:58:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76b4d4a58d606e268fb87d75152a1d46e02c0211","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 19:09:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/796331cf26cf4832978989850ef5d951\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1a3b16e00694caea31cb6b65842491f : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ccdbcf474574bd9b29b33a78d4b3cda : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/900aafc3630a4724a0d0087df2a7c307 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/70b19c3481e04d5c843c7f12c6c9f22d : SUCCESS in 2m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/595128219ccb40a3aa6d7634230e8e24 : SUCCESS in 3m 16s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/6997d4ffb1eb458da268d67a560777a0 : SUCCESS in 3m 06s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/6614ba1e5c5244a3a65c33854c0b569c : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"b7e0c4236700ba53ded576569ca253ccc8315c22","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-02-19 21:14:06.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9bd1d66266b4e47d77c46b237ab09bc11434c99","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-02-19 21:45:10.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4ea340ab2bd183b829fc929e4e112fd9c90b875","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-02-19 21:59:23.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db1627625e50137bd9707ff13552bf994681dbfa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 21:59:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d4682be6c5e866140d1036f746955d8fb873b004","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 22:09:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60aa64f4c8e341618ac2135546c4186a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d65f2326c3034724af3143e6849ac393 : SUCCESS in 2m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dee4a33226f24873b4e90db56afe8d3b : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5fc83d7457a4a7289bc448647c5bdd3 : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85bdec23580541a18629a1ac0eaef17b : SUCCESS in 2m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/78c890bb028a433aa5e4758009bb9e69 : SUCCESS in 3m 18s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/bfb747bc2a0d46808b00b92d73756eb7 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"153b2790f1efe00f9b6c1eb7bea785fe9e9860b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 22:09:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c1929fdc44e237401ea8d3d81cc0507360e29c16","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 22:10:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/106b23aff8314c969eb287851e41f6d3\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/6d52c42cc64d4770bf42cfce375a4932 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c4cd58035f97d245f7eab8f3e4f713a8842f6bc","revisions":{"4c4cd58035f97d245f7eab8f3e4f713a8842f6bc":{"kind":"REWORK","_number":1,"created":"2025-02-19 18:58:53.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/34/942234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/34/942234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/34/942234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/34/942234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/34/942234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/34/942234/1"}}},"commit":{"parents":[{"commit":"15fdf66868090e109c2469cb9990f6754b52a5e3","subject":"Merge \"Adding Ghanshyam candidacy for QA PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/15fdf66868090e109c2469cb9990f6754b52a5e3"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-02-19 18:58:34.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-02-19 18:58:34.000000000","tz":60},"subject":"Add Artem Goncharov candidacy for TC","message":"Add Artem Goncharov candidacy for TC\n\nChange-Id: I72f011dc897cd9bd92885961238b043559d431bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/4c4cd58035f97d245f7eab8f3e4f713a8842f6bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/4c4cd58035f97d245f7eab8f3e4f713a8842f6bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
