)]}'
{"id":"openstack%2Felection~942408","triplet_id":"openstack%2Felection~master~Ib55919624653fec8765208b8896fa0dae3ae4779","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2025-03-02 15:56:50.000000000","reason":"Change was abandoned"},"30623":{"account":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"last_update":"2025-03-02 15:56:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib55919624653fec8765208b8896fa0dae3ae4779","subject":"Add Suzhengwei candidancy for Masakari PTL","status":"ABANDONED","created":"2025-02-21 01:12:59.000000000","updated":"2025-03-02 15:56:50.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4275715ce7727cb35fed3cbb283b057bb2d685c3","_number":942408,"virtual_id_number":942408,"owner":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-21 01:21:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":-1,"date":"2025-02-21 01:38:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-21 01:21:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-21 01:38:27.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2025-02-21 20:55:45.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"}],"messages":[{"id":"34c10fad9effcf0ee1bc4ad2328134e9c04e2460","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"date":"2025-02-21 01:12:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c804cef8e3d6eb85bf970e79fba331f78b177f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 01:21:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/236b898e1f9149398b77d7bbf4d25919\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26fe70fb44ac461d9fb6dd91b2667af6 : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2c96c579f99414d88ccbd303bb8075b : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4e43b31b18d4564b86a8e4b9066e3e4 : SUCCESS in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e2802b33c2c446cace2aac32d443a86 : SUCCESS in 2m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/977e74683bcb4015b332d304ee637028 : SUCCESS in 3m 25s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/feffb00f1e4f4438b23e79e36e3ae6e5 : FAILURE in 3m 21s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/69093d3c2acb493fa9c69fd41094843e : FAILURE in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"51ca2aea2cbca67a415cbc7d4a46c96274a7f2e5","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-02-21 01:38:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a0793de73c4c6c7649ba25287e990ea0a93bc2a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-02-21 20:55:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4275715ce7727cb35fed3cbb283b057bb2d685c3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-03-02 15:56:50.000000000","message":"Abandoned\n\n2025.2 election was closed.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"513c399ecee67c0143cca883e43b9d56520802c7","revisions":{"513c399ecee67c0143cca883e43b9d56520802c7":{"kind":"REWORK","_number":1,"created":"2025-02-21 01:12:59.000000000","uploader":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"ref":"refs/changes/08/942408/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/08/942408/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/08/942408/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/08/942408/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/08/942408/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/08/942408/1"}}},"commit":{"parents":[{"commit":"4c4cd58035f97d245f7eab8f3e4f713a8842f6bc","subject":"Add Artem Goncharov candidacy for TC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/4c4cd58035f97d245f7eab8f3e4f713a8842f6bc"}]}],"author":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2025-02-21 01:12:41.000000000","tz":480},"committer":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2025-02-21 01:12:41.000000000","tz":480},"subject":"Add Suzhengwei candidancy for Masakari PTL","message":"Add Suzhengwei candidancy for Masakari PTL\n\nChange-Id: Ib55919624653fec8765208b8896fa0dae3ae4779\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/513c399ecee67c0143cca883e43b9d56520802c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/513c399ecee67c0143cca883e43b9d56520802c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
