)]}'
{"id":"openstack%2Felection~948494","triplet_id":"openstack%2Felection~master~I1b6838c1b923c6a8210ddf666cee1ca199c3c131","project":"openstack/election","branch":"master","topic":"2026.1-elections","attention_set":{},"removed_from_attention_set":{"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2025-05-12 07:43:34.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-05-12 07:43:34.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-05-12 07:30:56.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-05-12 07:43:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1b6838c1b923c6a8210ddf666cee1ca199c3c131","subject":"Add configuration for 2026.1/\"G\" elections","status":"MERGED","created":"2025-04-29 22:27:03.000000000","updated":"2025-05-12 07:44:18.000000000","submitted":"2025-05-12 07:43:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"948494-2026.1-elections","meta_rev_id":"44f9b50d01dd7ac88fd47cc4abe46eb3a73a59cf","_number":948494,"virtual_id_number":948494,"owner":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-12 07:43:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2025-05-11 22:25:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2025-05-12 07:30:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2025-05-12 07:30:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-29 22:49:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-29 23:02:14.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2025-04-30 07:51:14.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2025-05-02 18:07:07.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2025-05-09 19:42:59.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-05-12 07:30:56.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"76764a9c63bfadcc67cb638e3fbf37cdbd7b5306","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-04-29 22:27:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26d1aedf4c686b8bb0ab39ecc81938551907feca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 22:49:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/371035ee55c64a659603d0c51b9ed641\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daca72aa23ec4a898dd6c2d8f89476c3 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a6550d1cfb942a08e1b7a12376b2d23 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92b15326d740416d90c431bc905f227d : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/761d0427e5d94192917487d1d0054d5f : SUCCESS in 2m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/527e58a4c7d349528a450cb4a66cf6ad : SUCCESS in 2m 58s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/c4b44a74ccfc42d09f40af7c907bb75e : SUCCESS in 2m 54s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/26d19de4b21248eaab64010ee7556ec0 : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":1},{"id":"bd4c9508edc450325854acec5188ca4a11f3c900","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-04-29 23:02:14.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"df8ce1703eb644160a129224c252a757b97df16d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-04-30 07:51:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24936cadf13d0e95cfd0734e40204cd920602c06","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-04-30 15:49:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1b6ba83e80337cd4146e21e1bf662aedd210ad51","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-04-30 19:25:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dba899dc60abdffe1c2dfab30a16541c9b10ba1a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-05-02 18:07:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3cc61d481f2b81ed65cf42ebce540f3282263d18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-05-09 17:48:20.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f6c444e6a133c72ddf0706bee8135e5545e7aa94","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-05-09 17:50:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fb42c240094e43e6b05653cc0ef07506a2098287","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-05-09 17:55:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e3c5bca9ff5ad62d83d87c55a73e24ae00a9edd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 17:55:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/40697e3565ae42a2b32333ff5aadc883\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34c76a5cc7474e24a6285c78727bb6cc : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebb0a08f05ff4b9586a9e6ec376d35a3 : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c87fb26b19a4441bf16e7a2b21c88a5 : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68ad99367a074a9ba3ec6e7c9e37f83a : SUCCESS in 2m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/193faeebb769440c9fcea3a3c6e55cc5 : FAILURE in 2m 28s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/2c0d90fc46f14e0b8c723c0d07de8c10 : SUCCESS in 2m 53s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/3d1b8c0d28dd491babc6c11bb23dc6fc : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":2},{"id":"1af5404901f0554e0fb18ec6ab725158b21a2ce6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-05-09 19:42:59.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9318c99d97be2b82824f461cbdf2e4f57d5df123","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-05-09 20:01:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca363a02a699e641cb9f4fb61bee13a7e71dc002","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-05-10 00:34:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"46456466dc81660df0e1e68a398dad9ff914f45a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-10 00:37:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a870762eecc4b1492b37e94eb6d7997\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67924b2073664a8c823dff051aca27b3 : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7abd28a9345c4c8cb5b2b6b2bdbae33c : SUCCESS in 1m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2bd3211e2e946cd96798afe978ac9f9 : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d4ad575296c4638b6e53255beadc391 : SUCCESS in 1m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/691216aff1394abb8a868bff18171b73 : FAILURE in 2m 29s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/46a7a64a646c4922a5a33548f030aedf : SUCCESS in 1m 38s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/1b4ac27702884f2294d0c01d5ea61cc4 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":3},{"id":"cbee9a4192af4c9e83682bd8115e02eb7c729f53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-05-10 00:52:19.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"88a3ea69972aab716c0d468b3f67eb0d96c9a7c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-10 00:55:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d49281eb37f47fbb2e4f7550b3b1671\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ab7a31376ec4ec38b449698d8cd6eb3 : SUCCESS in 1m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c98b118aa344674a7db1c660970ee33 : SUCCESS in 1m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b39b72c593b495b9d0e2d505184a188 : SUCCESS in 1m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c13e960d535f4bc9b580dca63509d93e : SUCCESS in 1m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/005da83e0bca4b28a4d83a9adca1c7bf : SUCCESS in 1m 39s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/a9b90eea75d24bf3a89131f7bb8f865a : SUCCESS in 2m 31s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/3374bf1b63f64ad98f3f685ede550458 : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":4},{"id":"44e8d966b2633f0f64cf43fc44a2cb171e445140","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-05-10 01:10:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c804874ed9f82effc62d700f1b00f8e063acb75f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-05-11 22:25:09.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b009273f0d5c61aec9ab26351aa0fda7e648ad80","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-05-12 07:30:56.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2c8a3bfb68ccc7a787a802ca8170ebe58359cf6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 07:31:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c7ea3ddb1205ce5c4c96a8d5cd83006b9b65b36e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 07:43:33.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3f17866272a41d8a7be380196d2720c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e91643ed4166413fbd8a637daae46e5e : SUCCESS in 1m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e200ad0721c441f2bb0c98dea9f93cf3 : SUCCESS in 1m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22c8be066447416298343bc9c32702b6 : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/031ae9b6846e470396ecca8241ad0f5b : SUCCESS in 4m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0f4a8db3f28e4f878432499b6e924f1c : SUCCESS in 2m 51s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/23e3d0a9a697419aa46de91e4b237737 : SUCCESS in 2m 13s","accounts_in_message":[],"_revision_number":4},{"id":"c1a419824c73ec17a19ad9b58e4342cb337f66e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 07:43:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"44f9b50d01dd7ac88fd47cc4abe46eb3a73a59cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 07:44:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12dce5c1a71b48adb48c06b0563f6249\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/bd9b904aa66146bda19ccfc752ae02a6 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"978c8f2eac8deec254853dfc7f621e014044adb2","revisions":{"d1d6bab7c437ed997300ddaedd4aa53ba9799b15":{"kind":"REWORK","_number":1,"created":"2025-04-29 22:27:03.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/94/948494/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/94/948494/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/94/948494/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/94/948494/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/94/948494/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/94/948494/1"}}},"commit":{"parents":[{"commit":"e1785c84c96f2abe88e6ee390a7ff6cc8f675b9f","subject":"Merge \"Close 2025.2 Election Results (TC/PTL)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e1785c84c96f2abe88e6ee390a7ff6cc8f675b9f"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-04-29 22:25:03.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-04-29 22:25:03.000000000","tz":540},"subject":"Add configuration for 2026.1/\"G\" elections","message":"Add configuration for 2026.1/\"G\" elections\n\nThis is the result of the execution from \"setup-election-config\"\ncommand on the repo assuming release date as 2025-10-01\nper current release information on https://releases.openstack.org.\n\nPaste log:\nhttps://paste.opendev.org/show/bds7A2cvkI8c9h1Hq4Go/\n\nChange-Id: I1b6838c1b923c6a8210ddf666cee1ca199c3c131\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d1d6bab7c437ed997300ddaedd4aa53ba9799b15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d1d6bab7c437ed997300ddaedd4aa53ba9799b15"}]},"branch":"refs/heads/master"},"c90f9ae59b1f9fb4354055e1051f3ea369c33d45":{"kind":"REWORK","_number":2,"created":"2025-05-09 17:48:20.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/94/948494/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/94/948494/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/94/948494/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/94/948494/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/94/948494/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/94/948494/2"}}},"commit":{"parents":[{"commit":"e1785c84c96f2abe88e6ee390a7ff6cc8f675b9f","subject":"Merge \"Close 2025.2 Election Results (TC/PTL)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e1785c84c96f2abe88e6ee390a7ff6cc8f675b9f"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-04-29 22:25:03.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-05-09 17:48:20.000000000","tz":0},"subject":"Add configuration for 2026.1/\"G\" elections","message":"Add configuration for 2026.1/\"G\" elections\n\nThis is the result of the execution from \"setup-election-config\"\ncommand on the repo assuming release date as 2025-10-01\nper current release information on https://releases.openstack.org\nand setting voting weeks as 3 rather than default 2 from TC suggestion.\n\nPaste log:\nhttps://paste.opendev.org/show/bidSFRdkbuKYNSS2yXGm/\n\nChange-Id: I1b6838c1b923c6a8210ddf666cee1ca199c3c131\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c90f9ae59b1f9fb4354055e1051f3ea369c33d45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c90f9ae59b1f9fb4354055e1051f3ea369c33d45"}]},"branch":"refs/heads/master"},"12e128d500489ca9e545d395b794e573014af01f":{"kind":"REWORK","_number":3,"created":"2025-05-10 00:34:04.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/94/948494/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/94/948494/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/94/948494/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/94/948494/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/94/948494/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/94/948494/3"}}},"commit":{"parents":[{"commit":"e1785c84c96f2abe88e6ee390a7ff6cc8f675b9f","subject":"Merge \"Close 2025.2 Election Results (TC/PTL)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e1785c84c96f2abe88e6ee390a7ff6cc8f675b9f"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-04-29 22:25:03.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-05-10 00:33:50.000000000","tz":540},"subject":"Add configuration for 2026.1/\"G\" elections","message":"Add configuration for 2026.1/\"G\" elections\n\nThis is the result of the execution from \"setup-election-config\"\ncommand on the repo assuming release date as 2025-10-01\nper current release information on https://releases.openstack.org\nand setting voting weeks as 3 rather than default 2 from TC suggestion.\n\nPaste log:\nhttps://paste.opendev.org/show/bidSFRdkbuKYNSS2yXGm/\n\nChange-Id: I1b6838c1b923c6a8210ddf666cee1ca199c3c131\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/12e128d500489ca9e545d395b794e573014af01f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/12e128d500489ca9e545d395b794e573014af01f"}]},"branch":"refs/heads/master"},"978c8f2eac8deec254853dfc7f621e014044adb2":{"kind":"REWORK","_number":4,"created":"2025-05-10 00:52:19.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/94/948494/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/94/948494/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/94/948494/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/94/948494/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/94/948494/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/94/948494/4"}}},"commit":{"parents":[{"commit":"e1785c84c96f2abe88e6ee390a7ff6cc8f675b9f","subject":"Merge \"Close 2025.2 Election Results (TC/PTL)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e1785c84c96f2abe88e6ee390a7ff6cc8f675b9f"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-04-29 22:25:03.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-05-10 00:52:09.000000000","tz":540},"subject":"Add configuration for 2026.1/\"G\" elections","message":"Add configuration for 2026.1/\"G\" elections\n\nThis is the result of the execution from \"setup-election-config\"\ncommand on the repo assuming release date as 2025-10-01\nper current release information on https://releases.openstack.org\nand setting voting weeks as 3 rather than default 2 from TC suggestion.\n\nPaste log:\nhttps://paste.opendev.org/show/bidSFRdkbuKYNSS2yXGm/\n\nChange-Id: I1b6838c1b923c6a8210ddf666cee1ca199c3c131\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/978c8f2eac8deec254853dfc7f621e014044adb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/978c8f2eac8deec254853dfc7f621e014044adb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
