)]}'
{"id":"openstack%2Felection~957035","triplet_id":"openstack%2Felection~master~I205bec32dfbfd0888133d0543f6f02c8932a2886","project":"openstack/election","branch":"master","topic":"ptl-zaqar-2026.1","attention_set":{},"removed_from_attention_set":{"8846":{"account":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"last_update":"2025-08-13 22:42:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I205bec32dfbfd0888133d0543f6f02c8932a2886","subject":"Add Hao Wang candidacy for Zaqar 2026.1 PTL","status":"MERGED","created":"2025-08-11 17:31:27.000000000","updated":"2025-08-13 22:43:22.000000000","submitted":"2025-08-13 22:42:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957035-ptl-zaqar-2026.1","meta_rev_id":"908e916a4ed96e977877929a356a184ce6faac30","_number":957035,"virtual_id_number":957035,"owner":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-13 22:42:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2025-08-13 22:12:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2025-08-12 06:31:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2025-08-13 22:12:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-11 17:41:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-12 06:31:53.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-08-13 22:12:23.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"72c986bbe45e982ed8a2be4c34dca378b00784de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-08-11 17:31:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e704988f88dc921cf46d09f3102a30d18f6c833","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 17:41:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80d933e168f842f6b44637dd33fa335a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/246e3d25fa144892ad108901b3d0054a : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38f186a4913e4f4e8caeafbcad5a7cee : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e38a03bfcca34d8eb6105c258eb4c766 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/749fbe2602e7470dad17f3a617cb96a4 : SUCCESS in 3m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b7ebeeb057fb471eadb820c683e8b8d6 : SUCCESS in 1m 45s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/1baae934f7754a17836c068883fa1d08 : SUCCESS in 3m 03s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/58c942517e7e4bfea3596ba8efa7a33f : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0cf4a82e3b2fb112f02a59bd5baf08b2398f49a4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-08-12 06:31:53.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"050189ded30d48a4303566d3ddb88a10da49f463","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-08-13 22:12:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"16615943a65d57ffd14aec8f090e6011635a401b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 22:13:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd687c2d638f14a071f4271387c309b29e1cb9d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 22:42:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bebbefaa79f454983c3bd03f66692e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ac92dea43ce4e7cbaf3bfb97cdacd4c : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04c811252917403e811641c922d6f4fa : SUCCESS in 2m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/391dc6f51f244f0092584710c32fedde : SUCCESS in 1m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/615d505baaa74782a6e6531c44717858 : SUCCESS in 2m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4cc01560b14a48ebbfffa089ae803713 : SUCCESS in 3m 07s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/7e96cf3a7f6d4583ba304543c5b9e3b6 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"0847c9bc59630b3c26e979acb5ff6cc89cd753bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 22:42:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"908e916a4ed96e977877929a356a184ce6faac30","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 22:43:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e155981d60a142f7955f3142032ee5b6\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/18eca86e173443f388e1b293d395d63e : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"723d8e1d8ce3418a9223dcfec6dfbc00fb4f2fe1","revisions":{"723d8e1d8ce3418a9223dcfec6dfbc00fb4f2fe1":{"kind":"REWORK","_number":1,"created":"2025-08-11 17:31:27.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/35/957035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/35/957035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/35/957035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/35/957035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/35/957035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/35/957035/1"}}},"commit":{"parents":[{"commit":"5cb68b79fe9364c773f6404341101fcd9ad8a747","subject":"Merge \"Use newer OpenInfra member profile URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/5cb68b79fe9364c773f6404341101fcd9ad8a747"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-08-11 17:30:12.000000000","tz":-420},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-08-11 17:30:45.000000000","tz":-420},"subject":"Add Hao Wang candidacy for Zaqar 2026.1 PTL","message":"Add Hao Wang candidacy for Zaqar 2026.1 PTL\n\nChange-Id: I205bec32dfbfd0888133d0543f6f02c8932a2886\nSigned-off-by: Hao Wang \u003csxmatch1986@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/723d8e1d8ce3418a9223dcfec6dfbc00fb4f2fe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/723d8e1d8ce3418a9223dcfec6dfbc00fb4f2fe1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
