)]}'
{"id":"openstack%2Felection~957152","triplet_id":"openstack%2Felection~master~I817c1e9dce0e21556af8c44fc604d8a7f297f3d0","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"6282":{"account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"last_update":"2025-08-13 22:42:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I817c1e9dce0e21556af8c44fc604d8a7f297f3d0","subject":"Adding wu.wenxiang candidacy for skyline","status":"MERGED","created":"2025-08-12 15:57:42.000000000","updated":"2025-08-13 22:44:05.000000000","submitted":"2025-08-13 22:42:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957152","meta_rev_id":"175c245cc374c57aa51140f974b7f8852277bea2","_number":957152,"virtual_id_number":957152,"owner":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-13 22:42:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2025-08-13 22:17:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2025-08-13 08:44:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2025-08-13 22:17:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-12 16:07:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-13 08:44:49.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-08-13 22:17:08.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"588391c332dd92e12b8e4092fe249cc14474c381","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-08-12 15:57:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6dc913c5535e9d4372b3f8e76f73e7db162979c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 16:07:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8156d03bbf5c4888b6c4b3d654142268\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20afb9d8961540288fc141dbde28a81c : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0391d3609f2c4ae194831a8b8ebb7912 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a93aeefe9fcd48e581e6dd7bbb458041 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89f13ee6af304c22b70310c2d327d12f : SUCCESS in 3m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9e407278e844409f84ab04a4042db674 : SUCCESS in 2m 42s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/4ce3a08c24284e2ea529c0112cf9dc16 : SUCCESS in 2m 40s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/726b5d7861e64d96907b238689ba887b : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"7537f80448f3271845d81d21776b45a57932aeea","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-08-13 08:44:49.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f92ef15e2322d82c8754f542594ee316fd4a92c6","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-08-13 22:17:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b3fdcd6929eb3936869639f0feaeec354c6fc7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 22:17:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02749d0aeeb570402f0808dcf9b5d6743adb8dbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 22:42:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1e573e88c6e45d1835720f4198b99d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cede6729256c4511b1433b9423b41b75 : SUCCESS in 2m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9dce9f6e192649b1834713e01e73a199 : SUCCESS in 1m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28dbb6c4624c4ffebfdca593136caa1b : SUCCESS in 2m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bb5536709e24acb833f886d939697f1 : SUCCESS in 2m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1427444ee6ed4589a89ef31fa087891e : SUCCESS in 2m 12s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/e8df1c14a62c414aaf8a3c83a48fa00f : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"cea6feffc9c3fd42267a020a82d60f3360f4156d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 22:42:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"175c245cc374c57aa51140f974b7f8852277bea2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 22:44:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/652def153fad45cf9c0fe17514082d95\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/bd4851a4435f4e5aa79be64b753a77cb : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87303e6af4d66835603ddd491cdcef761b950102","revisions":{"87303e6af4d66835603ddd491cdcef761b950102":{"kind":"REWORK","_number":1,"created":"2025-08-12 15:57:42.000000000","uploader":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"ref":"refs/changes/52/957152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/52/957152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/52/957152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/52/957152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/52/957152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/52/957152/1"}}},"commit":{"parents":[{"commit":"5cb68b79fe9364c773f6404341101fcd9ad8a747","subject":"Merge \"Use newer OpenInfra member profile URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/5cb68b79fe9364c773f6404341101fcd9ad8a747"}]}],"author":{"name":"Wu Wenxiang","email":"wu.wenxiang@99cloud.net","date":"2025-08-12 15:55:49.000000000","tz":480},"committer":{"name":"Wu Wenxiang","email":"wu.wenxiang@99cloud.net","date":"2025-08-12 15:57:20.000000000","tz":480},"subject":"Adding wu.wenxiang candidacy for skyline","message":"Adding wu.wenxiang candidacy for skyline\n\nSigned-off-by: Wu Wenxiang \u003cwu.wenxiang@99cloud.net\u003e\nChange-Id: I817c1e9dce0e21556af8c44fc604d8a7f297f3d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/87303e6af4d66835603ddd491cdcef761b950102"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/87303e6af4d66835603ddd491cdcef761b950102"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
