)]}'
{"id":"openstack%2Felection~957240","triplet_id":"openstack%2Felection~master~I4415c33d3745d80334c9334ae558ffb471e66f5b","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2025-08-13 22:18:52.000000000","reason":"\u003cGERRIT_ACCOUNT_14482\u003e replied on the change","reason_account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-08-18 07:58:00.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2025-08-18 09:29:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4415c33d3745d80334c9334ae558ffb471e66f5b","subject":"Adding René Ribaud candidacy for Nova 2026.1 PTL","status":"MERGED","created":"2025-08-13 13:39:35.000000000","updated":"2025-08-18 09:30:17.000000000","submitted":"2025-08-18 09:29:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957240","meta_rev_id":"7a46ab17fc924392d78c325763f5127ffcb210c6","_number":957240,"virtual_id_number":957240,"owner":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-18 09:29:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2025-08-13 22:18:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2025-08-18 07:58:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2025-08-18 07:58:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-13 13:44:55.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2025-08-13 13:44:55.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-08-13 17:27:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"65c16fef19b3da0d0de19ebf8a49a7d8aa90d3f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-08-13 13:39:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3594627d700728dfc060a9d18b6c44082ca84ccf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 17:27:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3226c272e6b42d88783bef20457291b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67c5d48b088c4c3a8f3e599f04141ea4 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/543630f2ad55492a8108a11a21c1d48b : SUCCESS in 1m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e21252335e70434aa9adfb562e35e53c : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f79b0560a494b878ae562b08e9db47a : SUCCESS in 4m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/59c37578c0d24cc3a5c6b5a722be1da8 : SUCCESS in 3m 01s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/26796fb3fc9a40f78b09a8a43f63e9fc : SUCCESS in 3m 09s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/e429a141201a4e9d80f8e162ea3e1990 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"553a9c25d8851d47e192805684d0b6b7e98c2e75","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-08-13 22:18:52.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f8668c2db4c4f8d804928341709471647ca58d4b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-08-18 07:58:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"38cc2d3c381aa0f2b9ddad0120893719b0a000c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 07:58:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc957018afc17e6c4c59b20d8b195654e53a1330","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 09:29:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ebee7d6908349e6a66cb8e85093cea1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2768cc1b27e45b78b701b7067b99ee7 : SUCCESS in 2m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f128f8b801394d8bb2de3ee543c2bb63 : SUCCESS in 1m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b8284ea7ae5480c8c8f061ff8ddc5a0 : SUCCESS in 1m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8a23efc6be94056a17149fc7614dafd : SUCCESS in 3m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8e56449b182b4366bb8f832887df79ec : SUCCESS in 4m 55s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/2349badc83894288a9ddf05db7f183b8 : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":1},{"id":"00550b0d968da4ad881de5eb8c29e63927bcfb96","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 09:29:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7a46ab17fc924392d78c325763f5127ffcb210c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 09:30:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b0fc649227b43b591abd6ea566e7b64\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/995b64ea6ccf4f99bf3f5998962ea9bd : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d91f2449e7fce0937f4dad893d14b7c28d7433e0","revisions":{"d91f2449e7fce0937f4dad893d14b7c28d7433e0":{"kind":"REWORK","_number":1,"created":"2025-08-13 13:39:35.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/40/957240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/40/957240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/40/957240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/40/957240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/40/957240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/40/957240/1"}}},"commit":{"parents":[{"commit":"5cb68b79fe9364c773f6404341101fcd9ad8a747","subject":"Merge \"Use newer OpenInfra member profile URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/5cb68b79fe9364c773f6404341101fcd9ad8a747"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2025-08-11 16:52:10.000000000","tz":120},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2025-08-13 13:38:25.000000000","tz":120},"subject":"Adding René Ribaud candidacy for Nova 2026.1 PTL","message":"Adding René Ribaud candidacy for Nova 2026.1 PTL\n\nChange-Id: I4415c33d3745d80334c9334ae558ffb471e66f5b\nSigned-off-by: René Ribaud \u003crene.ribaud@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d91f2449e7fce0937f4dad893d14b7c28d7433e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d91f2449e7fce0937f4dad893d14b7c28d7433e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
