)]}'
{"id":"openstack%2Felection~957658","triplet_id":"openstack%2Felection~master~Ib6f5a6bf44cfc13e393a785e179b5c13a1e93d5e","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-18 21:31:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib6f5a6bf44cfc13e393a785e179b5c13a1e93d5e","subject":"Adding Takashi Kajinami candidacy for Heat PTL","status":"MERGED","created":"2025-08-18 09:26:59.000000000","updated":"2025-08-18 21:32:44.000000000","submitted":"2025-08-18 21:31:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957658","meta_rev_id":"e0fc9c5a39dbff658d8ae2fb3af27bdf731fdfc2","_number":957658,"virtual_id_number":957658,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-18 21:31:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2025-08-18 21:19:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2025-08-18 12:09:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2025-08-18 21:19:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-18 09:39:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-18 12:09:30.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-08-18 21:19:01.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"168e93ddf03c5e1d6649874710726dc222df2a87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 09:26:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c782920ac277dbe928cac01191119ec14cc900c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 09:39:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0796e81fc20d4672ba6f0ee28bed1692\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de3cfd4846cb45869b2529e090af01a6 : FAILURE in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7354f221bf64cffaf4bc68770254626 : SUCCESS in 2m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2142d7ee1344717abaef87967e6dd8e : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2dfae6f89c034be1b689ad902fd8014f : SUCCESS in 4m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6ec475d5ce504e83ba7e3d62b2cb11be : SUCCESS in 4m 55s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/770a8eb578324af382463b4845f8eacb : FAILURE in 4m 57s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/eb1d75306a394b53a25a3ea466ac9c03 : FAILURE in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c355a81611b4de6d56b490aebf1a4d33181e37f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 11:24:45.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b1cf95cbce50cad8fcd3e732c1b257f9c0f2bb04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 11:37:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0acecc4a281c4eca8dfadba1a46a6a6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47bc767ac31d4c26916813d1622b5747 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d40430c37e745ca967af692268f0fcc : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/582c1384b8134f609686e8c59ad8ae6c : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/736d903d9a2647a29bdf417c6014e379 : SUCCESS in 1m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3cfd43c36655424e8f2f19713199716a : SUCCESS in 3m 27s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/95003c2c4d4b461c957ec9c4602a5c36 : SUCCESS in 1m 47s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/ba36a74a13ec4e3a98b2b4ed2863bd70 : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":2},{"id":"51081e9c8a6b286328761c13c861e279341d599d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-08-18 12:09:30.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cd8fdd6ac7270b13034e117ba56855294acc464a","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-08-18 21:19:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c3690ea33fe4b52ff0adbb96779cbf7da0e3d5fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 21:19:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8bbc254d4fdefea64ca093d6f6a7886b1158b793","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 21:31:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52d2509465df4c91bf084e49a65320aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5dd42a0f7af4c00ba81996429d1935f : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/925f074891164648bad9e7d607fae290 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac2cf6c1d6fc456bb5c5ea8529ebde4b : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/136ee65f0ff64da290a4192436eb8e54 : SUCCESS in 2m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ecc2964621a94d8d99feff2f0ba70abd : SUCCESS in 3m 49s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/b2c42ab722ee4732aaf2d4ed73e7056c : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":2},{"id":"51d04522a6a598403a319beaaa4534b4df63c8e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 21:31:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e0fc9c5a39dbff658d8ae2fb3af27bdf731fdfc2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 21:32:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/406bfe6aa3cc4674961bd9bd9d9c9fae\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/928acbb0227d431ca44fcdb27fabd0dd : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8ac049e78dcd3c9ce7750958fd20382a7879a071","revisions":{"e95629888614421130e7f362954bc08d063018c0":{"kind":"REWORK","_number":1,"created":"2025-08-18 09:26:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/957658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/58/957658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/58/957658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/58/957658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/58/957658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/58/957658/1"}}},"commit":{"parents":[{"commit":"87e52797e866bb9e08a4c7010ee7dade647a7616","subject":"Merge \"Add Jon Bernard candidacy for Cinder PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/87e52797e866bb9e08a4c7010ee7dade647a7616"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 09:26:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 09:26:51.000000000","tz":540},"subject":"Adding Takashi Kajinami candidacy for Heat PTL","message":"Adding Takashi Kajinami candidacy for Heat PTL\n\nChange-Id: Ib6f5a6bf44cfc13e393a785e179b5c13a1e93d5e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e95629888614421130e7f362954bc08d063018c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e95629888614421130e7f362954bc08d063018c0"}]},"branch":"refs/heads/master"},"8ac049e78dcd3c9ce7750958fd20382a7879a071":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-08-18 11:24:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/957658/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/58/957658/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/58/957658/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/58/957658/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/58/957658/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/58/957658/2"}}},"commit":{"parents":[{"commit":"2202febc40fea5401b7e6808ac3c9e6bd8127e76","subject":"Merge \"Adding Michael Johnson candidacy for Octavia\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2202febc40fea5401b7e6808ac3c9e6bd8127e76"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 09:26:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 11:24:45.000000000","tz":0},"subject":"Adding Takashi Kajinami candidacy for Heat PTL","message":"Adding Takashi Kajinami candidacy for Heat PTL\n\nChange-Id: Ib6f5a6bf44cfc13e393a785e179b5c13a1e93d5e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/8ac049e78dcd3c9ce7750958fd20382a7879a071"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/8ac049e78dcd3c9ce7750958fd20382a7879a071"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
