)]}'
{"id":"openstack%2Felection~957873","triplet_id":"openstack%2Felection~master~I781f8b9816c12152fbe6ffe52b3edca739ba94f3","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-08-20 08:04:03.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I781f8b9816c12152fbe6ffe52b3edca739ba94f3","subject":"Adding Matt Crees candidacy for Blazar","status":"ABANDONED","created":"2025-08-19 11:03:05.000000000","updated":"2025-08-20 08:04:03.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f3491084dbc78685edce69b24b048c2be53bd3fc","_number":957873,"virtual_id_number":957873,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-08-19 11:12:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":-1,"date":"2025-08-19 18:29:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":-1,"date":"2025-08-19 14:07:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-19 11:12:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-19 14:07:41.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-08-19 18:29:14.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"093d6f0064ce006e32ec40ed821b7432113abac1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-08-19 11:03:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2df7db6bfa8534cb8c904cdb2a8d2a5011195d44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 11:12:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab522de4884041ac93aad5c59f0b70b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47d64a0cb84c4888a031c1a5b5725063 : FAILURE in 2m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6273218416594959ab93524814184e27 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1eab5a4cc2164538aa6743cc9f032208 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f37a27dbc1754a6490ea8beaf4976d54 : SUCCESS in 4m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/06c209585f9d482f861a255422d039a1 : SUCCESS in 2m 45s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/7addead5d00a4df5b98cad3812889722 : FAILURE in 5m 32s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/56c8c1d8f4694fcd92f843a582029686 : FAILURE in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"6c0b42b796850b285501c71eac7e59a34502bf37","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-08-19 13:08:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0e1e2693e692f85c07e2cbbf11072a1fd001f00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 13:17:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0fb9e51766be44dbbfeb907e90b7c346\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a922b35c768486c9a56cffbab598400 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8dc982f3e8f74522b8c86f241161bb3f : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/843b1f7f2fe34e369c580d574e794b08 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1c9f8c7ec534981b66b600b41799feb : SUCCESS in 5m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7e20c90108f54180a546c64b41fa061c : SUCCESS in 2m 40s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/18a99a7311e34f74a2b4033d53ed2092 : FAILURE in 3m 34s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/2774c64ad77e4d6f98e2d25d1154f615 : FAILURE in 6m 35s","accounts_in_message":[],"_revision_number":1},{"id":"a3903b34fe55f31c0e8cb211abf93de4dde85b64","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-08-19 14:07:41.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f3377008b5cc1ebfcfe8872f57cb4d6c0116403","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-08-19 18:29:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f3491084dbc78685edce69b24b048c2be53bd3fc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-08-20 08:04:03.000000000","message":"Abandoned\n\nabandoned: I\u0027m ineligible","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f8dbdd09c4866c1c5e565099ae9ae7fb61fbc9c7","revisions":{"f8dbdd09c4866c1c5e565099ae9ae7fb61fbc9c7":{"kind":"REWORK","_number":1,"created":"2025-08-19 11:03:05.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/73/957873/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/73/957873/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/73/957873/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/73/957873/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/73/957873/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/73/957873/1"}}},"commit":{"parents":[{"commit":"5ff24c2d68af898fc16b51be94383454c052130a","subject":"Merge \"Adding Takashi Kajinami candidacy for Storlets PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/5ff24c2d68af898fc16b51be94383454c052130a"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-08-19 09:31:29.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-08-19 11:03:00.000000000","tz":60},"subject":"Adding Matt Crees candidacy for Blazar","message":"Adding Matt Crees candidacy for Blazar\n\nChange-Id: I781f8b9816c12152fbe6ffe52b3edca739ba94f3\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/f8dbdd09c4866c1c5e565099ae9ae7fb61fbc9c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/f8dbdd09c4866c1c5e565099ae9ae7fb61fbc9c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
