)]}'
{"id":"openstack%2Felection~957880","triplet_id":"openstack%2Felection~master~I2231c8aa4b95bcb56a0bc1e9496fea50e6c49d70","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-08-19 19:54:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2231c8aa4b95bcb56a0bc1e9496fea50e6c49d70","subject":"Adding Cyril Roelandt candidacy for Glance PTL 2026.1","status":"MERGED","created":"2025-08-19 11:57:03.000000000","updated":"2025-08-19 19:54:02.000000000","submitted":"2025-08-19 19:54:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957880","meta_rev_id":"2370ace402a4637ac76b8defe59b20a04b0d138d","_number":957880,"virtual_id_number":957880,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-19 19:54:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2025-08-19 18:25:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2025-08-19 14:04:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2025-08-19 12:34:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2025-08-19 18:25:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-19 12:11:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-19 12:20:04.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2025-08-19 12:34:55.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-08-19 14:04:22.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-08-19 18:25:05.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"d1b4dc3dd58dae1ba9a54438ced1d7a0e28e27a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-19 11:57:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2904e489f988ba054f494e7bd0b2d35724981bd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-19 12:00:18.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b56859d64a2a651f3704261d47514a5203bc14c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 12:11:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2d7669c1ec64c659bf52c7724d091df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b95ffe0ba224fc4bd5b5dc158e27565 : FAILURE in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05d2424219b346df8df926ad6a40da48 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d004efeca9304408a5d88498de35fc33 : SUCCESS in 2m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24b458ab27d7403ca5626c0e4bd8dc5b : SUCCESS in 2m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9c3dcfb8f6a747c88c596384fa8d3050 : SUCCESS in 4m 03s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/8f9470ed28f04d0f8f76818ef1a85f92 : FAILURE in 2m 39s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/ddfd7e323e944338bf791bca190082d7 : FAILURE in 2m 44s","accounts_in_message":[],"_revision_number":2},{"id":"67f22cc3866da094a6a674bc6a187a262e3aab69","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-08-19 12:20:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"16772e0ba191bf6204863b445666625cb0e56d9a","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-08-19 12:20:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"66d1930372e207646443444316fd214ab34810b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 12:30:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4db8939a2ac441cfbb437c36ba45c8c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69c524c5241d4594b1a8fe62944ac2a5 : SUCCESS in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba82a3a951344b27a021e95b1334a7ef : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7cde63e4e06b4feba0f1e3086f290089 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74d1ed4fd0464b1b87c21f3306977aff : SUCCESS in 2m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/af3cc9a0714348e0a5bcd17261ee720c : SUCCESS in 5m 23s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/a2e5bcd8d2154d84b25b7862dfe58e78 : SUCCESS in 2m 13s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/2935df28071b4108b78f03d122f13cdf : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":2},{"id":"b1b6f9f6ee78d23ec34cdae38026ae472e4ff812","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-08-19 12:34:55.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"992d5d946fc257f68ab087e5d9248c2950d3d8e0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-08-19 14:04:22.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"525266a7d62f9bd46aefa5bbf7613e680ec70154","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-08-19 18:25:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7831f0f7b1caaec0e05fcc4209bdfc9e93bb08b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 18:25:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1d66f2df8cb9438e93f8aa2e9fb1e96cc41011e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 19:54:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ebdc97d999b4944bac2c5f9a0902e09\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed962f182d9a4cb99d03857dded36ee6 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa6eca4bc6ee4927b3f291a0f5ec58da : SUCCESS in 2m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19b95802a4354015b8735f34565a255e : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/847e50cc08514b72be8a93b4efc3c1f0 : SUCCESS in 3m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/301559aa0c1c43fe99580d3b9e786404 : SUCCESS in 2m 41s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/405618bbf74a40d883a5a38cb51caf1f : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":2},{"id":"2370ace402a4637ac76b8defe59b20a04b0d138d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 19:54:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"76e24a94ba20c6470b04f6ce991c6fd999c25e90","revisions":{"977b84f960f0e3c2c018d7ec7b78e2255413d8d8":{"kind":"REWORK","_number":1,"created":"2025-08-19 11:57:03.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/80/957880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/80/957880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/80/957880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/80/957880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/80/957880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/80/957880/1"}}},"commit":{"parents":[{"commit":"5ff24c2d68af898fc16b51be94383454c052130a","subject":"Merge \"Adding Takashi Kajinami candidacy for Storlets PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/5ff24c2d68af898fc16b51be94383454c052130a"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-08-19 11:54:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-08-19 11:56:56.000000000","tz":0},"subject":"Adding Cyril Roelandt candidacy for Glance PTL 2026.1","message":"Adding Cyril Roelandt candidacy for Glance PTL 2026.1\n\nChange-Id: I2231c8aa4b95bcb56a0bc1e9496fea50e6c49d70\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/977b84f960f0e3c2c018d7ec7b78e2255413d8d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/977b84f960f0e3c2c018d7ec7b78e2255413d8d8"}]},"branch":"refs/heads/master"},"76e24a94ba20c6470b04f6ce991c6fd999c25e90":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-08-19 12:00:18.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/80/957880/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/80/957880/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/80/957880/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/80/957880/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/80/957880/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/80/957880/2"}}},"commit":{"parents":[{"commit":"5ff24c2d68af898fc16b51be94383454c052130a","subject":"Merge \"Adding Takashi Kajinami candidacy for Storlets PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/5ff24c2d68af898fc16b51be94383454c052130a"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-08-19 11:54:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-08-19 12:00:18.000000000","tz":0},"subject":"Adding Cyril Roelandt candidacy for Glance PTL 2026.1","message":"Adding Cyril Roelandt candidacy for Glance PTL 2026.1\n\nAs Cyril is currently on vacation, I am submitting this nomination\non his behalf after consultation with OpenStack election officials.\n\n\nChange-Id: I2231c8aa4b95bcb56a0bc1e9496fea50e6c49d70\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/76e24a94ba20c6470b04f6ce991c6fd999c25e90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/76e24a94ba20c6470b04f6ce991c6fd999c25e90"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
