)]}'
{"id":"openstack%2Felection~957940","triplet_id":"openstack%2Felection~master~Ife6bc727a2ad6527e16eaa9efcb424e28f2b2778","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-08-19 17:00:45.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-08-19 19:54:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ife6bc727a2ad6527e16eaa9efcb424e28f2b2778","subject":"Adding Arnaud Morin candidacy for Mistral","status":"MERGED","created":"2025-08-19 15:15:51.000000000","updated":"2025-08-19 19:55:28.000000000","submitted":"2025-08-19 19:54:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957940","meta_rev_id":"4e3a9064af8792b79795d75279a8e8c6cb258702","_number":957940,"virtual_id_number":957940,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-19 19:54:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2025-08-19 18:31:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2025-08-19 17:00:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2025-08-19 18:31:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-19 15:31:09.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"CC"},{"updated":"2025-08-19 15:31:09.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-08-19 15:53:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-19 18:31:35.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"3034287af9d9863ddff09ae0162e207e8e807bc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-08-19 15:15:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed12a66c2ffde5f1c331627645687d3072d424aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 15:53:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ab6c6f7f16948cd870ea32251ae2e30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bace9860b495487a89c952cfcfe064ca : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cb14fb430a64d528806719fdfd70e9f : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a6ed59062fb4d9c9d8153f69e0d28d8 : SUCCESS in 2m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb29c69eb6d44ce198b995f5666ce4eb : SUCCESS in 2m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e98a95be133a4cd4bd951fb8495301b1 : SUCCESS in 2m 08s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/2fd5c1bf15134a3e89dd0889bb54a2ac : SUCCESS in 2m 24s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/2e0d81fd719747fa9db07fc7a0cc1437 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":1},{"id":"434382be6f1f5c54a52cd58e4a6b2f5bcf975d5d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-08-19 17:00:45.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e13ae3c2dc7b0d5f6ba90602451795368ba1cec","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-08-19 18:31:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"abd5e8ef1949564da676aa15ad59adafdebfc65a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 18:31:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3338363a8e6b002e470d2b6acb369b77433802e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 19:54:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8933a3a9faee4d008f9ebb7b74dbfcac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59ccf40bfed34a97bb553d2c8c31fcf2 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d82f056e00348ba99d10c2f8312d422 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ccde58644414045a59c4d1915937fd0 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff6559adaf4046aaaa2ad5be8c937780 : SUCCESS in 4m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3a7f1b50692c4f039ae30f5eb8fd5442 : SUCCESS in 5m 26s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/1d90603a34ee425ca8717ef1e17e577d : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":1},{"id":"6dad3d1e992c747dd77f518d8beb10a5915ab589","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 19:54:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4e3a9064af8792b79795d75279a8e8c6cb258702","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 19:55:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/578c5866f7c141009b73ffceb14e44f0\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/72e9369bafbe4da98f4c391efd1251c9 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6a87f37ddb548fd5c39c34f0e767092cd21096d","revisions":{"d6a87f37ddb548fd5c39c34f0e767092cd21096d":{"kind":"REWORK","_number":1,"created":"2025-08-19 15:15:51.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/40/957940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/40/957940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/40/957940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/40/957940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/40/957940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/40/957940/1"}}},"commit":{"parents":[{"commit":"5ff24c2d68af898fc16b51be94383454c052130a","subject":"Merge \"Adding Takashi Kajinami candidacy for Storlets PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/5ff24c2d68af898fc16b51be94383454c052130a"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2025-08-19 15:13:21.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2025-08-19 15:15:47.000000000","tz":120},"subject":"Adding Arnaud Morin candidacy for Mistral","message":"Adding Arnaud Morin candidacy for Mistral\n\nChange-Id: Ife6bc727a2ad6527e16eaa9efcb424e28f2b2778\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d6a87f37ddb548fd5c39c34f0e767092cd21096d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d6a87f37ddb548fd5c39c34f0e767092cd21096d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
