)]}'
{"id":"openstack%2Felection~958991","triplet_id":"openstack%2Felection~master~I82e8be5a016d624d8e8486ae4953c68bc9994249","project":"openstack/election","branch":"master","topic":"direct_reminder_template","attention_set":{},"removed_from_attention_set":{"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2025-09-25 10:26:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I82e8be5a016d624d8e8486ae4953c68bc9994249","subject":"Add email template for direct reminder about CIVS","status":"MERGED","created":"2025-08-31 08:21:17.000000000","updated":"2025-09-25 10:27:37.000000000","submitted":"2025-09-25 10:26:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958991-direct_reminder_template","meta_rev_id":"1c9736dfffc7df18c953c09a1f3f4c5f9a158821","_number":958991,"virtual_id_number":958991,"owner":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-25 10:26:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2025-09-02 14:34:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2025-09-02 12:09:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2025-09-25 10:13:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-31 08:36:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-02 12:09:47.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-09-02 14:34:15.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"1f457f848ae1c146d4c2604af7f0bdcf29c171b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-08-31 08:21:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73709e97dbb79e531ce5ebd2278e15015eede377","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-08-31 08:22:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f702fb0c08ec1e4f0450419857fd5fe60a497750","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-08-31 08:23:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"783261f0b675d041f5e0b327600f9d6246ea983a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-08-31 08:26:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6f6eed53d1e333230259a1534b7caf77a114f2ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-31 08:36:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb496510a628409c906b2dd299421015\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/425d0638f0394bb19b5af69c05d54d81 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6acfe042351344e2892c96ce065a269c : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71529fd7af5747a4bc2b5c1e7b3de3a4 : SUCCESS in 2m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b729a3990474a19812256a8f909ed2c : SUCCESS in 3m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bd6ea484b33f4b169e154066dc961cb5 : SUCCESS in 5m 09s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/e5f2ed9aa2be4f1ca0aef9319763733d : SUCCESS in 3m 28s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/761a92c78e7a4a23864b5decb5f0d562 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":4},{"id":"452a808643d472fda16b5589e117f810cbcbe376","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-09-02 12:09:47.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"5726f256cd413b1a169644c228e589c71c86bdd3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-02 14:34:15.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bf1eccd1d493f93a8462c618fe41fb8f670c45ed","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-09-25 10:13:11.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1cb58b60a801ff2b6ed19f59db3f287647535cf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 10:13:26.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"16c1060c6d6c66ddc47a99f358105bc103b2817f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 10:26:04.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc17c47ca551463fb43ca79203ecc874\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46a60a1765d74e50883ee3e31453385b : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c71976cbb772439d90fa82049ec9c406 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53135c9ab19644d8ab389dccac421da5 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/001a100223304bc8a61f0197dbf7bd0d : SUCCESS in 3m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4501f1c0cc7a44288e43dd4e8a545e9f : SUCCESS in 3m 05s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/29f0e52bbc344ba0a00d730259c0a44a : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":4},{"id":"45c744ff8f32ac97dd22c07f53afc4642ef52485","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 10:26:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"1c9736dfffc7df18c953c09a1f3f4c5f9a158821","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 10:27:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54deb7ac1e384a1eb3b00b9473146d8e\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/bd78e7146f4440ef8892c5cba6b2460c : SUCCESS in 34s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fc6cc9ea685af4d1e2031d16bab2569ae50ef9be","revisions":{"6640c835f4d55c4065932cb02a9461ef282ff163":{"kind":"REWORK","_number":1,"created":"2025-08-31 08:21:17.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/91/958991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/91/958991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/91/958991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/91/958991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/91/958991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/91/958991/1"}}},"commit":{"parents":[{"commit":"a4c97b7a7bd236c7bae4c1ca2115767d6aad3c86","subject":"Merge \"Tim Burke candidacy for Swift PTL (Gazpacho)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/a4c97b7a7bd236c7bae4c1ca2115767d6aad3c86"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-08-31 07:40:21.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-08-31 08:21:02.000000000","tz":540},"subject":"Add email template for direct reminder about CIVS","message":"Add email template for direct reminder about CIVS\n\nThis new template can be generated for a single project or tc\n(by specifying project as \u0027tc\u0027 for combined).\nIt is intended to be sent during voting periods to the contributors\nwho are not activated to CIVS voting system.\n\nChange-Id: I82e8be5a016d624d8e8486ae4953c68bc9994249\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/6640c835f4d55c4065932cb02a9461ef282ff163"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/6640c835f4d55c4065932cb02a9461ef282ff163"}]},"branch":"refs/heads/master"},"135c1484c007d483d0588ee12a1e4a18650cc5dc":{"kind":"REWORK","_number":2,"created":"2025-08-31 08:22:20.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/91/958991/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/91/958991/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/91/958991/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/91/958991/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/91/958991/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/91/958991/2"}}},"commit":{"parents":[{"commit":"a4c97b7a7bd236c7bae4c1ca2115767d6aad3c86","subject":"Merge \"Tim Burke candidacy for Swift PTL (Gazpacho)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/a4c97b7a7bd236c7bae4c1ca2115767d6aad3c86"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-08-31 07:40:21.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-08-31 08:22:10.000000000","tz":540},"subject":"Add email template for direct reminder about CIVS","message":"Add email template for direct reminder about CIVS\n\nThis new template can be generated for a single project or tc\n(by specifying project as \u0027tc\u0027 for combined).\nIt is intended to be sent during voting periods to the contributors\nwho are not activated to CIVS voting system.\n\nChange-Id: I82e8be5a016d624d8e8486ae4953c68bc9994249\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/135c1484c007d483d0588ee12a1e4a18650cc5dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/135c1484c007d483d0588ee12a1e4a18650cc5dc"}]},"branch":"refs/heads/master"},"b93bc0ebed4c47d81b28542ebae48dde6634887f":{"kind":"REWORK","_number":3,"created":"2025-08-31 08:23:07.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/91/958991/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/91/958991/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/91/958991/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/91/958991/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/91/958991/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/91/958991/3"}}},"commit":{"parents":[{"commit":"a4c97b7a7bd236c7bae4c1ca2115767d6aad3c86","subject":"Merge \"Tim Burke candidacy for Swift PTL (Gazpacho)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/a4c97b7a7bd236c7bae4c1ca2115767d6aad3c86"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-08-31 07:40:21.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-08-31 08:22:57.000000000","tz":540},"subject":"Add email template for direct reminder about CIVS","message":"Add email template for direct reminder about CIVS\n\nThis new template can be generated for a single project or tc\n(by specifying project as \u0027tc\u0027 for combined).\nIt is intended to be sent during voting periods to the contributors\nwho are not activated to CIVS voting system.\n\nChange-Id: I82e8be5a016d624d8e8486ae4953c68bc9994249\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/b93bc0ebed4c47d81b28542ebae48dde6634887f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/b93bc0ebed4c47d81b28542ebae48dde6634887f"}]},"branch":"refs/heads/master"},"fc6cc9ea685af4d1e2031d16bab2569ae50ef9be":{"kind":"REWORK","_number":4,"created":"2025-08-31 08:26:02.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/91/958991/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/91/958991/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/91/958991/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/91/958991/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/91/958991/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/91/958991/4"}}},"commit":{"parents":[{"commit":"a4c97b7a7bd236c7bae4c1ca2115767d6aad3c86","subject":"Merge \"Tim Burke candidacy for Swift PTL (Gazpacho)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/a4c97b7a7bd236c7bae4c1ca2115767d6aad3c86"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-08-31 07:40:21.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-08-31 08:25:50.000000000","tz":540},"subject":"Add email template for direct reminder about CIVS","message":"Add email template for direct reminder about CIVS\n\nThis new template can be generated for a single project or tc\n(by specifying project as \u0027tc\u0027 for combined).\nIt is intended to be sent during voting periods to the contributors\nwho are not activated to CIVS voting system.\n\nChange-Id: I82e8be5a016d624d8e8486ae4953c68bc9994249\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/fc6cc9ea685af4d1e2031d16bab2569ae50ef9be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/fc6cc9ea685af4d1e2031d16bab2569ae50ef9be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
