)]}'
{"id":"openstack%2Felection~961776","triplet_id":"openstack%2Felection~master~Ia36027269da51fb2d7e0223daef9fcbf642a8b52","project":"openstack/election","branch":"master","topic":"fix-update-governance-script","attention_set":{},"removed_from_attention_set":{"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2025-09-23 22:38:48.000000000","reason":"\u003cGERRIT_ACCOUNT_14482\u003e replied on the change","reason_account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-09-23 22:47:28.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-09-23 22:47:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia36027269da51fb2d7e0223daef9fcbf642a8b52","subject":"Fix update-governance script","status":"MERGED","created":"2025-09-19 07:40:51.000000000","updated":"2025-09-23 22:48:12.000000000","submitted":"2025-09-23 22:47:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961776-fix-update-governance-script","meta_rev_id":"93421bf3aff5586e6e82525ba4163fa970938728","_number":961776,"virtual_id_number":961776,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-23 22:47:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2025-09-23 22:38:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2025-09-22 16:32:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2025-09-23 22:38:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-19 08:04:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-19 08:49:08.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-09-19 22:16:22.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2025-09-22 16:32:34.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"789e3a796cbb0b0b9cec0fa4dba0774b9187fb55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-09-19 07:40:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43d62b9035eef6773dd3e2b3c21bf89a953c76b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 08:04:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b46f6f221d24804aecfabcd5fccbd69\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b178be1fe0cc4e27a421fdb91af04e7c : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b50db4722ede426a8412a4ed01d398c8 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/341eaf694ac843dfa5770193b33919f0 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6daacd6416c94b28b09f2ba359aac50b : SUCCESS in 3m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a6d61c1bf1ce4bb3b2184a18859d35d8 : SUCCESS in 3m 01s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/44628d3b12a144cf9664a37cff7873a5 : SUCCESS in 4m 46s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/fe89dd5a035540afb1d944883be8ade6 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"c11266b14ce3caf55689cad9449783a361a4ab94","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-19 08:49:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1a188a0f818a75fa92bbf73cadb71eae23f635d6","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-09-19 22:16:22.000000000","message":"Patch Set 1: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"11a7da3a897ce5c73020153b864a1779e0301d18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-09-22 07:35:33.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f9d7e3f313d38281880083a2acc63b3a04c8eb80","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-09-22 07:35:45.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"54641b2c6c4bbcbfbd9f8a411776a7d2b9ad60de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 07:50:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/021a09893ccc42649aa06075c7e7071b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e44746607298468eadab9bf0c15ed953 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7206dab00814db8a29885bac9061682 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7f5c99db5e8421fb0a2126cbc0e20c4 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ec90439cda2452e98b156f4a87127cc : SUCCESS in 3m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/65cdd4421c3b4a85a4d706aa0a19bbf9 : SUCCESS in 2m 57s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/47aa82c8a6024089b8d767d6e47a2b79 : SUCCESS in 3m 26s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/ac55837e3b5648de80fda9c7104f9411 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":2},{"id":"568410f06e105c8d44d0d3daefa0861b5a29abe2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-09-22 08:23:40.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8c3a3f0f63214f1cdab2bcbd8945f4760a038c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 08:35:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cb1d3fbe69248428599e0ad8ec9a4b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8969bdfe56f34714a67dcbfdb04845f0 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/928c6e4b9eb54dbea1ff524d3235f43d : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/184c44595f8a466f95fb14ec28e81ff8 : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1175f43fe12741e9bed68d8975c5f7d6 : SUCCESS in 4m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/77cd3002872141f782964027dc6a7439 : SUCCESS in 3m 30s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/b7f35d1404cf492597d8fd70a0131819 : SUCCESS in 3m 18s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/17f720bf7766449f8a4d8de84976aab3 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":3},{"id":"f50a888b5a09e02d823b6c672307d7e914fdf380","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-22 16:32:34.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"299438c3ca18b7adf7909012a78f479479ee33b8","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-09-23 22:38:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b4fc5ae4db74d1024baef28e00baa5375580cc49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 22:38:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"aa7de2c42ab3754a4375bae69222a7f3738e66fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 22:47:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3dbebd604c947aa943bb2e898284a09\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8da7a65a4d0b413cb99a5a5e18ea85bf : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beea4beb36a944e1acc3c5f83c516f84 : SUCCESS in 2m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8de3d758de024bacbd034f600050937c : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1dd109ec368245f5810986e3076f44a3 : SUCCESS in 3m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3be392564abf45eab5eb452afbeda3a4 : SUCCESS in 2m 43s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/3864f87e37af41b28827f639f4261f5b : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":3},{"id":"a1c7c5b5ddca92e6976557d51fd54b2f70ced323","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 22:47:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"93421bf3aff5586e6e82525ba4163fa970938728","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 22:48:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/229104613cda48cbaab756972c13f3c3\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/4aeee23b4f5445b4ba8f73553a641f35 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5695b355212b9439a563fc3ee3b128b94a8017d1","revisions":{"50a9f57441251c011e8503e6429b552130121425":{"kind":"REWORK","_number":1,"created":"2025-09-19 07:40:51.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/76/961776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/76/961776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/76/961776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/76/961776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/76/961776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/76/961776/1"}}},"commit":{"parents":[{"commit":"3b2ef67394de2f65194ce3747498beeb37099327","subject":"Merge \"docs: clarify OIF renewal requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3b2ef67394de2f65194ce3747498beeb37099327"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-09-19 07:39:45.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-09-19 07:39:45.000000000","tz":120},"subject":"Fix update-governance script","message":"Fix update-governance script\n\nPreviously it was always taking first candidate from the PTL candidates\nlist without checking which one was actually elected. Now it is looking\nfor the one with `elected` set to `true`.\n\nChange-Id: Ia36027269da51fb2d7e0223daef9fcbf642a8b52\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/50a9f57441251c011e8503e6429b552130121425"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/50a9f57441251c011e8503e6429b552130121425"}]},"branch":"refs/heads/master"},"c972b21bbdeb7f0320f611b2695a0f07f46b9051":{"kind":"REWORK","_number":2,"created":"2025-09-22 07:35:33.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/76/961776/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/76/961776/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/76/961776/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/76/961776/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/76/961776/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/76/961776/2"}}},"commit":{"parents":[{"commit":"3b2ef67394de2f65194ce3747498beeb37099327","subject":"Merge \"docs: clarify OIF renewal requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3b2ef67394de2f65194ce3747498beeb37099327"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-09-19 07:39:45.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-09-22 07:35:32.000000000","tz":0},"subject":"Fix update-governance script","message":"Fix update-governance script\n\nPreviously it was always taking first candidate from the PTL candidates\nlist without checking which one was actually elected. Now it is looking\nfor the one with `elected` set to `true`.\n\nChange-Id: Ia36027269da51fb2d7e0223daef9fcbf642a8b52\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c972b21bbdeb7f0320f611b2695a0f07f46b9051"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c972b21bbdeb7f0320f611b2695a0f07f46b9051"}]},"branch":"refs/heads/master"},"5695b355212b9439a563fc3ee3b128b94a8017d1":{"kind":"REWORK","_number":3,"created":"2025-09-22 08:23:40.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/76/961776/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/76/961776/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/76/961776/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/76/961776/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/76/961776/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/76/961776/3"}}},"commit":{"parents":[{"commit":"3b2ef67394de2f65194ce3747498beeb37099327","subject":"Merge \"docs: clarify OIF renewal requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3b2ef67394de2f65194ce3747498beeb37099327"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-09-19 07:39:45.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-09-22 08:23:22.000000000","tz":120},"subject":"Fix update-governance script","message":"Fix update-governance script\n\nPreviously it was always taking first candidate from the PTL candidates\nlist without checking which one was actually elected. Now it is looking\nfor the one with `elected` set to `true`.\n\nChange-Id: Ia36027269da51fb2d7e0223daef9fcbf642a8b52\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/5695b355212b9439a563fc3ee3b128b94a8017d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/5695b355212b9439a563fc3ee3b128b94a8017d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
