)]}'
{"id":"openstack%2Felection~975718","triplet_id":"openstack%2Felection~master~Ia3936d90b747d9738c7c3a04bfa68d964d78fd17","project":"openstack/election","branch":"master","topic":"2026.2-elections","attention_set":{},"removed_from_attention_set":{"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2026-02-04 23:55:43.000000000","reason":"removed on reply"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-02-05 13:47:15.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-02-05 07:47:15.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-02-05 13:47:15.000000000","reason":"Change was submitted"}},"hashtags":["2026.2-elections","dpl-reset"],"change_id":"Ia3936d90b747d9738c7c3a04bfa68d964d78fd17","subject":"Add nomination directories for Oslo and Requirements","status":"MERGED","created":"2026-02-04 23:45:29.000000000","updated":"2026-02-05 13:48:11.000000000","submitted":"2026-02-05 13:47:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975718-2026.2-elections","meta_rev_id":"e54b620a7aba2814b22554eb9259d31fe5a7e72b","_number":975718,"virtual_id_number":975718,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-05 13:47:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2026-02-04 23:55:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2026-02-05 07:47:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2026-02-05 13:35:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-04 23:46:45.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2026-02-04 23:46:45.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-02-04 23:46:45.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-02-04 23:59:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a7d68c98b5683300657ea61388e19d64bad19070","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-04 23:45:29.000000000","message":"Hashtags added: 2026.2-elections, dpl-reset","accounts_in_message":[],"_revision_number":1},{"id":"9f15b1af86919ce3f09fa516982d16fc10117487","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-04 23:46:13.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1ef1ac38a8b5b2c4ef7ba7bb37b32a4e1e4c1bb0","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2026-02-04 23:55:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ecc2ac52c20a00211f4d4bd117a1e81a636a4308","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 23:59:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4e01532d8ce46edb5517d3bf5a311d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecb457729db84c20aba0e5399941b85d : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d0243d561694a6b8c3736b57706745b : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c624d2f6b0c4cfc958a79bda8db8e5d : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07ba9ff82fa34a1897cee6106c326a80 : SUCCESS in 3m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ea9d38fa5a0d42efbd363cccf09a0100 : SUCCESS in 3m 02s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/d0c893ba76cd40aaa2b11c0859c08e21 : SUCCESS in 5m 20s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/3dc62ccf5f554538a0bfa41ee7e3f27d : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":2},{"id":"963a431078e71a34ef651131e0f21143107f9fee","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-05 07:47:15.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ffb4c98d39d915d8776115671eef4244175fe4b1","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2026-02-05 13:35:08.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d9737dc07d95b6444e7a04f6391a9fa5571707c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 13:35:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"97f63963b918ca86804b5ab5b0939e18d4508133","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 13:47:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cb6d3addda6416694c306bd8772e49c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/906cb3aa702d48be81c67a9bbb10bc73 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/064d24ac7d284cdfabaa62ad63fb6b25 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/012324f60fed4f16bef109f3fb89bd14 : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a5586ffb0e14fefabdbba026a364d6e : SUCCESS in 3m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0931e544004b44e09b55d3478ad24b38 : SUCCESS in 3m 30s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/62f10ebf833d442ba9947a1a6fbe517f : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":2},{"id":"2a5ba6a4dd4de1acc1ff34f75f35150c40c1e2f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 13:47:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e54b620a7aba2814b22554eb9259d31fe5a7e72b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 13:48:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00981892989548848b003a197e5fbf88\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/ad23314d36264072808534b86dcd7405 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c385593ee34ae11d6c804f678ad7375ad036606a","revisions":{"33e7c5ddf646a9a822ca9de0f532c60cb2f626f8":{"kind":"REWORK","_number":1,"created":"2026-02-04 23:45:29.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/18/975718/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/18/975718/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/18/975718/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/18/975718/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/18/975718/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/18/975718/1"}}},"commit":{"parents":[{"commit":"c3572a25abca5b97796a23b5332e6b85ccb4de93","subject":"Add watcher nomination dir for 2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c3572a25abca5b97796a23b5332e6b85ccb4de93"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-02-04 23:43:33.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-02-04 23:43:33.000000000","tz":-480},"subject":"Add nomination directories for Oslo and Requirements","message":"Add nomination directories for Oslo and Requirements\n\nWe\u0027ve had no response from one or more DPL liaisons\nin these teams [1][2] to continue using the DPL model\nfor the 2026.2 release cycle. So per procedure, we\u0027ll\naccept PTL nominations for these teams.\n\nChange-Id: Ia3936d90b747d9738c7c3a04bfa68d964d78fd17\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/33e7c5ddf646a9a822ca9de0f532c60cb2f626f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/33e7c5ddf646a9a822ca9de0f532c60cb2f626f8"}]},"branch":"refs/heads/master"},"c385593ee34ae11d6c804f678ad7375ad036606a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-02-04 23:46:13.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/18/975718/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/18/975718/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/18/975718/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/18/975718/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/18/975718/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/18/975718/2"}}},"commit":{"parents":[{"commit":"c3572a25abca5b97796a23b5332e6b85ccb4de93","subject":"Add watcher nomination dir for 2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c3572a25abca5b97796a23b5332e6b85ccb4de93"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-02-04 23:43:33.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-02-04 23:45:49.000000000","tz":-480},"subject":"Add nomination directories for Oslo and Requirements","message":"Add nomination directories for Oslo and Requirements\n\nWe\u0027ve had no response from one or more DPL liaisons\nin these teams [1][2] to continue using the DPL model\nfor the 2026.2 release cycle. So per procedure, we\u0027ll\naccept PTL nominations for these teams.\n\n[1] https://review.opendev.org/c/openstack/governance/+/974927\n[2] https://review.opendev.org/c/openstack/governance/+/974929\nChange-Id: Ia3936d90b747d9738c7c3a04bfa68d964d78fd17\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c385593ee34ae11d6c804f678ad7375ad036606a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c385593ee34ae11d6c804f678ad7375ad036606a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
