)]}'
{"id":"openstack%2Felection~976976","triplet_id":"openstack%2Felection~master~Ie85085df1161822d73e73a99eb352a121181c85a","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-16 16:07:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie85085df1161822d73e73a99eb352a121181c85a","subject":"Add Ivan Anfimov candidacy for Barbican PTL","status":"MERGED","created":"2026-02-16 14:16:31.000000000","updated":"2026-02-16 16:08:25.000000000","submitted":"2026-02-16 16:07:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976976","meta_rev_id":"8bc2eb4ee77ce8f174b6f46de8d392e50e923c91","_number":976976,"virtual_id_number":976976,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-16 16:07:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2026-02-16 15:53:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2026-02-16 15:53:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-16 14:28:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-16 15:53:49.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"feafb63aa0c07e04636e2d2e7fe253ad03d5f2a8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-16 14:16:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7faf1bb15cbfc5bc489eec31931c11cedf31f40","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-16 14:19:00.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4412a6ed7f24b2a75bb45988ef3b9cce3090d272","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-16 14:19:04.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"166944becf0b29c5e9804e7f11381d42c73b5fc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 14:28:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22d723a840b24586be8f64bab7eba74e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01283b27911d47aa8129ecbb5d2c29f3 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3eb907654bd40a6968bec79609b3527 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b8f6abce15d4b72b10b76983d8d5c59 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/794bd2f461d545c7b6245fd1e3c3890b : SUCCESS in 3m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ffeac2f08fe64be599e2e8718084cf86 : SUCCESS in 2m 04s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/d9c533d091064ab8920750ce1176fb4e : SUCCESS in 3m 35s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/02fc7eb9e33f4fd9a2fdeef45c55561c : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"c5e6aea57edad612a5a0cca282d7a1406d5b0134","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-02-16 15:53:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"91c713ebb6f7c514e348fe1a5b43821288ac8fc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 15:54:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"69433507ef0d44f67fc81d025894d578c09083a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 16:07:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abcbfc2f5ba3424294cfb7a51c1fbd9c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/012315464e3b42c3955953c7099b5ffa : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c772f0d03d1429dbc854d25f62b6dc5 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f962d46eb35c491894edc70f45b14b9b : SUCCESS in 2m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00724fc2105c4053b9cfc7c78501857a : SUCCESS in 3m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/788b09e6afe94e14b5da691435d71f08 : SUCCESS in 3m 41s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/a87d3fd44cc24523b74331be330b4bd7 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":2},{"id":"9e8d3a5b95b45864f7c28d4813b926589887fb7e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 16:07:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8bc2eb4ee77ce8f174b6f46de8d392e50e923c91","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 16:08:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/604d1f2e04e34192990551e971a9f0ff\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/a1c5dcb749e6413b99deab702babb5c7 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"afe333f4d0e8be56c6de28dd43e3f1d4f9d5fe83","revisions":{"4bcb6e4209d34507cec3d9fce8f327b9fc4096b6":{"kind":"REWORK","_number":1,"created":"2026-02-16 14:16:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/76/976976/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/76/976976/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/76/976976/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/76/976976/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/76/976976/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/76/976976/1"}}},"commit":{"parents":[{"commit":"c7dab1d2adb6ae78392e0f2f8c1cb0d99773c7f7","subject":"Merge \"Add Brian Haley (haleyb) candidacy for Neutron PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c7dab1d2adb6ae78392e0f2f8c1cb0d99773c7f7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-16 14:16:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-16 14:16:31.000000000","tz":0},"subject":"Add Ivan Anfimov candidacy for Barbican PTL","message":"Add Ivan Anfimov candidacy for Barbican PTL\n\nChange-Id: Ie85085df1161822d73e73a99eb352a121181c85a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/4bcb6e4209d34507cec3d9fce8f327b9fc4096b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/4bcb6e4209d34507cec3d9fce8f327b9fc4096b6"}]},"branch":"refs/heads/master"},"afe333f4d0e8be56c6de28dd43e3f1d4f9d5fe83":{"kind":"REWORK","_number":2,"created":"2026-02-16 14:19:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/76/976976/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/76/976976/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/76/976976/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/76/976976/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/76/976976/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/76/976976/2"}}},"commit":{"parents":[{"commit":"c7dab1d2adb6ae78392e0f2f8c1cb0d99773c7f7","subject":"Merge \"Add Brian Haley (haleyb) candidacy for Neutron PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c7dab1d2adb6ae78392e0f2f8c1cb0d99773c7f7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-16 14:16:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-16 14:19:00.000000000","tz":0},"subject":"Add Ivan Anfimov candidacy for Barbican PTL","message":"Add Ivan Anfimov candidacy for Barbican PTL\n\nChange-Id: Ie85085df1161822d73e73a99eb352a121181c85a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/afe333f4d0e8be56c6de28dd43e3f1d4f9d5fe83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/afe333f4d0e8be56c6de28dd43e3f1d4f9d5fe83"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
