)]}'
{"id":"openstack%2Felection~977179","triplet_id":"openstack%2Felection~master~If5599ffbedac6327560788bacbc228320ed2722b","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-02-18 09:26:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If5599ffbedac6327560788bacbc228320ed2722b","subject":"Tim Burke candidacy for Swift PTL (Hibiscus)","status":"MERGED","created":"2026-02-18 02:14:44.000000000","updated":"2026-02-18 09:26:46.000000000","submitted":"2026-02-18 09:26:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977179","meta_rev_id":"b1107a133dc4d2070c4c4a8f3e4f6ed34b30ad14","_number":977179,"virtual_id_number":977179,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-18 09:26:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2026-02-18 09:04:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2026-02-18 09:04:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-18 02:57:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-18 08:52:36.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-02-18 09:04:22.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"e70b0aacc1ab204aa7b966446fc0aa4896b78d70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-02-18 02:14:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b2ef91bc9b94d009ff967d8ebd494a0f817beef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 02:57:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/113471b0c29a4d949898133926fb6059\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/730acb46bd3e43c398290380b7a355bc : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0bab06a5c0c4aff91e97e8ca1abb19a : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e72a756d3b945059e71ef513b09532d : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6193783014b441b2ae40bc67cc160571 : FAILURE in 3m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0d83346b24564fc6b85db8010d4a6652 : SUCCESS in 4m 24s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/57147ac03cd04c04b3aece2ec3420e55 : SUCCESS in 4m 00s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/24260b4d2ccf4f8f87b8536dc93e0cc4 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"8f91d055c9dd1b28dd630cf0f97a817e8b9b0d9a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-18 08:52:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c49c0d03b21ba81c902790e38308dcc4b6a7f00","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-02-18 09:04:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"11de8cc49ea34fe688cfb7f4f6969fbbe5f272e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 09:11:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fef9aaad102e4a7a9d803f32baeeca74\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f279b86a105f48a98262c33e925f08a6 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38f0e77f3cf14a81a43c42830bb917b8 : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80cf2651fb4c4d7d939a85dba182dc2f : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7703d874802b45fe91876af487900d8e : SUCCESS in 3m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/edcd230003c348649a9487d98e0d5cb0 : SUCCESS in 2m 09s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/77d34498100e440f83a50cafc06a5818 : SUCCESS in 2m 32s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/449268c53e9246558585fb211921cbc7 : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9cdcbdfeeea30a4a1fa6291e6969e24a354f20c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 09:12:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c254a941ea07647320815827cfe3cb664b96803c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 09:26:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/071eee5c9d084f1293420c3195acb782\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6d4b58a1f6f4393886d81b00bd7d44f : SUCCESS in 2m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e45d0d42a8f4f2db7793e75d7a4f54b : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b086a62f3abd40c5917410c4c361da26 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dde749930ba444768cb4e5f27cef24f8 : SUCCESS in 3m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/49ec04857d864128a4370195d48f9577 : SUCCESS in 3m 10s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/f14afa333fb14edd9ca6711f616d9cbb : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b1107a133dc4d2070c4c4a8f3e4f6ed34b30ad14","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 09:26:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac32c2e25c1d5a5d1acaeea1b79bb487ec6d87be","revisions":{"ac32c2e25c1d5a5d1acaeea1b79bb487ec6d87be":{"kind":"REWORK","_number":1,"created":"2026-02-18 02:14:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/79/977179/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/79/977179/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/79/977179/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/79/977179/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/79/977179/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/79/977179/1"}}},"commit":{"parents":[{"commit":"540315d8ed063269c8b709ce6c990324ac4f7465","subject":"Merge \"Adding Mauricio Harley candidacy for Barbican\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/540315d8ed063269c8b709ce6c990324ac4f7465"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-02-18 02:14:04.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-02-18 02:14:39.000000000","tz":-480},"subject":"Tim Burke candidacy for Swift PTL (Hibiscus)","message":"Tim Burke candidacy for Swift PTL (Hibiscus)\n\nChange-Id: If5599ffbedac6327560788bacbc228320ed2722b\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ac32c2e25c1d5a5d1acaeea1b79bb487ec6d87be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ac32c2e25c1d5a5d1acaeea1b79bb487ec6d87be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
