)]}'
{"id":"openstack%2Felection~977187","triplet_id":"openstack%2Felection~master~I8f7aa7ca48cf3e5e79004103d7bec6f417dca22a","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-02-18 09:26:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8f7aa7ca48cf3e5e79004103d7bec6f417dca22a","subject":"Adding Erkin Mussurmankulov candidacy for Trove","status":"MERGED","created":"2026-02-18 07:08:27.000000000","updated":"2026-02-18 09:26:48.000000000","submitted":"2026-02-18 09:26:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977187","meta_rev_id":"0e5add38a950e5a1d5bedd3ae1abe862d090ee7b","_number":977187,"virtual_id_number":977187,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-18 09:26:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2026-02-18 09:15:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2026-02-18 09:15:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-18 07:18:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-18 09:15:19.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"8f4aa16419092d65328a2a989eeeefb91aa04e18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-02-18 07:08:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f658fd0283040794a476a45caf1d648a36d630a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 07:18:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a82d63e98c804e4896901a08aea6578c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/793df8073f4649509f9f450928adccbf : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/920de364fab045af87ce6c4dfd06c357 : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8d9c9e7d8d94112a3f4976629fbb9f4 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3cee90590e6a468fa479be758f0dba99 : FAILURE in 2m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/52c4e533e6574751895aa14d8d02423b : SUCCESS in 3m 09s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/dd838b3878a1479fb9e37263d4469862 : SUCCESS in 3m 32s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/6bb553541f7242eeafef0b849ac1a674 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"1482081932a7c47879a467c52251dbf1e1df5aa3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-02-18 08:25:09.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"97056c0f41ebe057599e3d019822f581048865ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 08:39:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7d9566822c84fcb8c9d2eb7f28f165e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6101a22f356d46a08e3c8baa97ece8b8 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5a6b0c9f4f246ae9a4a8996a729ec00 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9463d8696bba43f48e849168740b1106 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8534076c74b5446fb249bc1145c2bce4 : SUCCESS in 3m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9fc25a63433b4bd896fb5951ed9b2e0b : SUCCESS in 3m 29s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/09db078c4008423b8eceddba767d0bd1 : SUCCESS in 3m 32s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/a370307b38934bdf98ee44d2d925d48d : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":2},{"id":"f2c590e1b74f6242158f0d77de44f9204043b952","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-02-18 09:15:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3202e3322c78df9b39010ab3a880c389a64bca28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 09:15:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"26151e6d1c589419f10e90a0228ebb0ffd6b3b26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 09:26:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2da03da43f4f4b9f971dc25520553746\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a32211cbe44d469180e11344d6d58479 : SUCCESS in 2m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e51f6010f06e43d1b5b212deef686a39 : SUCCESS in 2m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86dcfc1112294128b3571ecf811f1e1b : SUCCESS in 2m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/299d38e887944eafa175992d11ebf591 : SUCCESS in 2m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a29b821fefdc4973a57ddaf4a44df9c0 : SUCCESS in 1m 59s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/3359286e494541aaba15b69525882d11 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":2},{"id":"0e5add38a950e5a1d5bedd3ae1abe862d090ee7b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 09:26:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c03dbd23cd8c65c494e0be43ab712fcff85cc13d","revisions":{"bd40d515361c2b21d8859c47d65449aa0b28eb29":{"kind":"REWORK","_number":1,"created":"2026-02-18 07:08:27.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/87/977187/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/87/977187/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/87/977187/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/87/977187/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/87/977187/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/87/977187/1"}}},"commit":{"parents":[{"commit":"540315d8ed063269c8b709ce6c990324ac4f7465","subject":"Merge \"Adding Mauricio Harley candidacy for Barbican\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/540315d8ed063269c8b709ce6c990324ac4f7465"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-02-17 07:42:03.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-02-17 08:02:46.000000000","tz":300},"subject":"Adding Erkin Mussurmankulov candidacy for Trove","message":"Adding Erkin Mussurmankulov candidacy for Trove\n\nChange-Id: I8f7aa7ca48cf3e5e79004103d7bec6f417dca22a\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/bd40d515361c2b21d8859c47d65449aa0b28eb29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/bd40d515361c2b21d8859c47d65449aa0b28eb29"}]},"branch":"refs/heads/master"},"c03dbd23cd8c65c494e0be43ab712fcff85cc13d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-02-18 08:25:09.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/87/977187/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/87/977187/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/87/977187/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/87/977187/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/87/977187/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/87/977187/2"}}},"commit":{"parents":[{"commit":"540315d8ed063269c8b709ce6c990324ac4f7465","subject":"Merge \"Adding Mauricio Harley candidacy for Barbican\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/540315d8ed063269c8b709ce6c990324ac4f7465"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-02-17 07:42:03.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-02-18 08:24:32.000000000","tz":300},"subject":"Adding Erkin Mussurmankulov candidacy for Trove","message":"Adding Erkin Mussurmankulov candidacy for Trove\n\nDepends-On: https://review.opendev.org/c/openstack/election/+/977160\nChange-Id: I8f7aa7ca48cf3e5e79004103d7bec6f417dca22a\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c03dbd23cd8c65c494e0be43ab712fcff85cc13d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c03dbd23cd8c65c494e0be43ab712fcff85cc13d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
