)]}'
{"id":"openstack%2Felection~977196","triplet_id":"openstack%2Felection~master~I45dccdb758908c56adb0d82dcc9d1a45966fc974","project":"openstack/election","branch":"master","topic":"stat-fix","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-02-19 18:00:40.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-06-22 23:53:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I45dccdb758908c56adb0d82dcc9d1a45966fc974","subject":"Fix render_statistics.py script","status":"MERGED","created":"2026-02-18 09:37:53.000000000","updated":"2026-06-22 23:54:50.000000000","submitted":"2026-06-22 23:53:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977196-stat-fix","meta_rev_id":"5a2e75f8bed4c2f7658a1566fc45762a4c3d6132","_number":977196,"virtual_id_number":977196,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-22 23:53:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2026-02-21 00:36:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2026-02-19 18:00:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2026-06-22 23:45:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-18 09:53:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-19 05:44:57.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-02-21 00:36:28.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"2107460d30a907ae6373d1de15a82f69c8dfe86b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-18 09:37:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"731260eb712d35c9c39e531fc22b9ede34ffc34f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 09:53:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e134189ab4c847adaab16b06b2198cc8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31bdce40424743be8e4ab973a1a4b6e7 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7c99abda3f646e697770661b4592173 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/836b7ddf4d1d43acb9e3a81d620a0674 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/983d2efb0fc246309aa7dcf4d2d48700 : SUCCESS in 2m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/71920f1715fe45fcb6ed520db1c676e0 : SUCCESS in 3m 23s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/8c8ee1d792154de685825c2f06d55690 : SUCCESS in 5m 50s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/0a9661f4463549d890103e826c5d0429 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5c7395c7c74458d1b0b6a0c6b159deeb8b37040b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-19 05:44:57.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"558ceb2d2d0615984a488f96bf2e846ed0a9cd8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-19 15:39:07.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"fdf37a93e8946cfeaf41b49a42d82cbe6539fbe2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-19 15:39:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0b2cdf4f1f02436e23a5b73df37bcdeec325b947","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 15:51:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8ace9d2f7714cf3ad42d8a6d7bc56ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22e77acb4d9a496c9877ff5cff094134 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45157922e4224adf8d8c55bab9997593 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c856b9fe00c49ee8e5c954cfc1f1a34 : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4eac5d6264c441568987a4be5b1a538d : SUCCESS in 3m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d2574c55512546ae927cb8191d488816 : SUCCESS in 3m 45s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/b38fe66b4d9d4801836cf38b5a15c7b8 : SUCCESS in 3m 15s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/ec091d1f2e1645d0888e920c1cf4d100 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":2},{"id":"35a98cf0c79ca5ca955ef16e64c8486bdafc4440","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-19 18:00:40.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2574e29f4106574719af8b12bd7a87d39e0926d9","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2026-02-21 00:36:28.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1c6b9957f00cfabf2231a0b825e6884f96f45e0f","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2026-06-22 23:45:18.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"381f988b34600d32280647dabe4feb01dd816773","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 23:45:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9072d2274fb95d7ef8b7baa5fde367ad38bda2b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 23:53:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86dd6ca3455d4af8bd564390e6dc7080\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/928138f6847449459371afbaf7a071cd : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77a50bcf762d45fb9992817a53488e4e : SUCCESS in 2m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/080706c53d5142c5b286c799d3093ee2 : SUCCESS in 3m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/74c2a7b6598a4258b7513700ffdfe5e7 : SUCCESS in 2m 03s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/e9ac67598d6e457ab955453ac72c30d1 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":2},{"id":"a5ed8767cc5fe8b3a7ef66283726863ec41c6de7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 23:53:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5a2e75f8bed4c2f7658a1566fc45762a4c3d6132","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 23:54:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92ebedc2853047239a0dd903e1e56ffa\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/94b6ea3c14af4fe8b837c5964a330e69 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e0ac8109faaa6eaddb41363edc97bde1014d4743","revisions":{"ced067c128016f6252c9a80f8030d3d38a4c4b0f":{"kind":"REWORK","_number":1,"created":"2026-02-18 09:37:53.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/96/977196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/96/977196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/96/977196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/96/977196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/96/977196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/96/977196/1"}}},"commit":{"parents":[{"commit":"211336f705af31e7e30d36a05478291198aa49ce","subject":"Merge \"Add Ivan Anfimov candidacy for Barbican PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/211336f705af31e7e30d36a05478291198aa49ce"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-02-18 09:34:34.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-02-18 09:34:34.000000000","tz":60},"subject":"Fix render_statistics.py script","message":"Fix render_statistics.py script\n\nInstead of replacing the list of candidates for a project with unmerged\ncandidacies, add a fake candidate. As a result, the script\u0027s output will\ncorrectly show the \"election needed\" status in case there is already an\napproved candidacy for the project in addition to the pending one.\n\nFix some wording along the way.\n\nChange-Id: I45dccdb758908c56adb0d82dcc9d1a45966fc974\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ced067c128016f6252c9a80f8030d3d38a4c4b0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ced067c128016f6252c9a80f8030d3d38a4c4b0f"}]},"branch":"refs/heads/master"},"e0ac8109faaa6eaddb41363edc97bde1014d4743":{"kind":"REWORK","_number":2,"created":"2026-02-19 15:39:07.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/96/977196/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/96/977196/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/96/977196/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/96/977196/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/96/977196/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/96/977196/2"}}},"commit":{"parents":[{"commit":"211336f705af31e7e30d36a05478291198aa49ce","subject":"Merge \"Add Ivan Anfimov candidacy for Barbican PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/211336f705af31e7e30d36a05478291198aa49ce"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-02-18 09:34:34.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-02-19 15:39:07.000000000","tz":0},"subject":"Fix render_statistics.py script","message":"Fix render_statistics.py script\n\nInstead of replacing the list of candidates for a project with unmerged\ncandidacies, add a fake candidate. As a result, the script\u0027s output will\ncorrectly show the \"election needed\" status in case there is already an\napproved candidacy for the project in addition to the pending one.\n\nFix some wording along the way.\n\nChange-Id: I45dccdb758908c56adb0d82dcc9d1a45966fc974\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e0ac8109faaa6eaddb41363edc97bde1014d4743"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e0ac8109faaa6eaddb41363edc97bde1014d4743"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
