)]}'
{"id":"openstack%2Felection~977258","triplet_id":"openstack%2Felection~master~I2e84fbe71b90a6f8f5ecc3b7dd5036e5264dc6bd","project":"openstack/election","branch":"master","topic":"charms-2026.2","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-02-18 21:03:16.000000000","reason":"Change was submitted"},"2424":{"account":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"last_update":"2026-02-18 21:03:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2e84fbe71b90a6f8f5ecc3b7dd5036e5264dc6bd","subject":"Adding Felipe Reyes for OpenStack Charms PTL","status":"MERGED","created":"2026-02-18 17:53:40.000000000","updated":"2026-02-18 21:04:36.000000000","submitted":"2026-02-18 21:03:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977258-charms-2026.2","meta_rev_id":"69a24c60eb68d53f508adfe7a42aa07d89be3356","_number":977258,"virtual_id_number":977258,"owner":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-18 21:03:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2026-02-18 17:54:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2026-02-18 20:54:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2026-02-18 20:54:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-18 17:54:37.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-02-18 18:03:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-18 20:54:28.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"a47e235817fe8af0e4f7caada73944a4bcb5ebb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2026-02-18 17:53:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6ae6a9926419674a294b3081063c6ed98b25557","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-18 17:54:37.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b22a9cdfaabe5aad55457809fc2d717fb834938a","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2026-02-18 18:02:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"33ed3b8e8b9c5006824b960071ee6ab1de09eb17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 18:03:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd7bbdc032d7409f9aec234ca8c6b6b5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4a59d6452b04cac8c2d92c276b7aca2 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ddd5bcae9044e239f003ae160fb71d2 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0699d32b87c04933a7457956d4dfcce4 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b7a8769494248fc9878bc3ead36ff99 : SUCCESS in 3m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4fb0d2992ef44e608817e55f15252bac : SUCCESS in 2m 00s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/44507546ee434442b51d0969e7284eb7 : SUCCESS in 2m 22s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/de20d4fb13a84148be2df6b4ca4e511b : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"b16c832cf4dbcb86ced013594af5b2f009165301","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-02-18 20:54:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f1bd97e6ea8a5b3229ee3297c97b64ddd1f7ae1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 20:54:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4155ade19e3c52664559b965edf542617cf6c43c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 21:03:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9adb083ab034842beefb071531ba68a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e17a4ecf8c04759b4a425fce7a323ad : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68753edc23af4fdca03b75454a94c423 : SUCCESS in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73390ec69a734365b5fbd0b2cc9818e9 : SUCCESS in 1m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba1440ba7277425fa180c2e26de9de0c : SUCCESS in 3m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fb47c62db7334c4f8ab418b9435384ec : SUCCESS in 4m 58s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/7fa73e2a20344275853931c2fc5a80c7 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":1},{"id":"5da2900684c530170ded1a9393dcffacc08f96e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 21:03:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"69a24c60eb68d53f508adfe7a42aa07d89be3356","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 21:04:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad51ff74d2a34178961624c5841902da\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/76062dd267f54082a83b71ab03b3be7d : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"45a99c3d5c66b7c24338e303fe9699d32b2d8f36","revisions":{"45a99c3d5c66b7c24338e303fe9699d32b2d8f36":{"kind":"REWORK","_number":1,"created":"2026-02-18 17:53:40.000000000","uploader":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"ref":"refs/changes/58/977258/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/58/977258/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/58/977258/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/58/977258/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/58/977258/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/58/977258/1"}}},"commit":{"parents":[{"commit":"168bb49fc803b9c90222150013449b3a20c63108","subject":"Merge \"Adding Rafael Weingärtner candidacy for CloudKitty\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/168bb49fc803b9c90222150013449b3a20c63108"}]}],"author":{"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","date":"2026-02-18 17:53:13.000000000","tz":-180},"committer":{"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","date":"2026-02-18 17:53:13.000000000","tz":-180},"subject":"Adding Felipe Reyes for OpenStack Charms PTL","message":"Adding Felipe Reyes for OpenStack Charms PTL\n\nChange-Id: I2e84fbe71b90a6f8f5ecc3b7dd5036e5264dc6bd\nSigned-off-by: Felipe Reyes \u003cfelipe.reyes@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/45a99c3d5c66b7c24338e303fe9699d32b2d8f36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/45a99c3d5c66b7c24338e303fe9699d32b2d8f36"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
