)]}'
{"id":"openstack%2Felection~977271","triplet_id":"openstack%2Felection~master~Ie22a85dbfb9e290d432103d7d83176b0ab067fc7","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-02-18 21:16:37.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-02-18 20:59:56.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}},"hashtags":[],"change_id":"Ie22a85dbfb9e290d432103d7d83176b0ab067fc7","subject":"Add Jon Bernard candidacy for Cinder PTL","status":"MERGED","created":"2026-02-18 20:27:09.000000000","updated":"2026-02-18 21:17:24.000000000","submitted":"2026-02-18 21:16:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977271","meta_rev_id":"df6f323925b58b86e20ceefffd38361e8453701c","_number":977271,"virtual_id_number":977271,"owner":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-18 21:16:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2026-02-18 20:59:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2026-02-18 20:59:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-18 20:36:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-18 20:56:04.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"db31a3474dba4b0086ddf91f4eb76252a1cc41b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-02-18 20:27:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88d4c5c7dbcbc8e0c545b8b07208757c0070b883","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 20:36:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eaaf865439c544468605c20df457f255\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3546fc2b23b04867b81990156a266596 : FAILURE in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af4f9a7c6108492bbd9ad27180af9b32 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5099621e38f7463c808cb409db8405af : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e8f482d343544f408e7a67d22a159995 : SUCCESS in 3m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/55f093a4b0ee4b9d878b28a62f11c854 : SUCCESS in 2m 54s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/0411e8bbcea34496a3e0e447cfc188a4 : FAILURE in 3m 02s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/f200f303f7b24cd3baba699fc3d177f3 : FAILURE in 2m 56s","accounts_in_message":[],"_revision_number":1},{"id":"dc2e225a79f8b22cdb7bde5fa595584fb79e859d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-02-18 20:56:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0008af1254b274b16727f287788773cc7b4e7fd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-02-18 20:59:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9f8ed9bfb84c595ee7c55a63da7c6cc218b0d35b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-02-18 20:59:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a46c7578cf78f93b567798d443c2b6b82bf06dee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 21:08:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/466626bc37a54e6dbfd5f22c13677bc3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/302d8e4d7bc64a319a66fa022accd0c1 : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5aa4365e1e5447d1a4670bba46bc416d : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3eeaf1c31b749cb84d618f66877dbbe : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b32755b1061447e9aa4844ebc7c9dba : SUCCESS in 3m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6a4be3bde273455babaf53997424280f : SUCCESS in 2m 01s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/49921138a23a49aca6bf56ada0b18c46 : SUCCESS in 4m 51s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/1b347651d25f4c71aa7e34671684658d : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":2},{"id":"b5da65b01d5252cca43a2f5fa8ee406a2d40bb8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 21:08:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a499476ab726aeda1fa89020c8516d83ab1cb825","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 21:16:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce8140a6ca9f4a9c92032ce9fd8c6d21\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae53a203251d4f3fb58a9e43199fcb5c : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38a9b54a567d4dc099d84d147ac5624c : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f1c94fe878f4b2d94ee367ff6cfd0a9 : SUCCESS in 2m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eaba3f6e008f462e8e11d7d8cd68e039 : SUCCESS in 4m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3678a4bc14ab4aae84b176b50ed7495d : SUCCESS in 4m 45s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/ef7445a9250944ea925bfd9edd2b77b3 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":2},{"id":"7e77040bb292360301bd4cecbba4ceceba264714","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 21:16:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"df6f323925b58b86e20ceefffd38361e8453701c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 21:17:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0439dad2fa854b8d83cc79a3ebf4c74c\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/1c1922497df5442fa2601ff1402fad0a : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"125ac1db1350b02012c9ffa74fe840fb38f7ec8d","revisions":{"412a3e6b6f518fda34b3ebe71aa9a5f2e173e655":{"kind":"REWORK","_number":1,"created":"2026-02-18 20:27:09.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/71/977271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/71/977271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/71/977271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/71/977271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/71/977271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/71/977271/1"}}},"commit":{"parents":[{"commit":"168bb49fc803b9c90222150013449b3a20c63108","subject":"Merge \"Adding Rafael Weingärtner candidacy for CloudKitty\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/168bb49fc803b9c90222150013449b3a20c63108"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2026-02-18 15:26:22.000000000","tz":-300},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2026-02-18 15:26:35.000000000","tz":-300},"subject":"Add Jon Bernard candidacy for Cinder PTL","message":"Add Jon Bernard candidacy for Cinder PTL\n\nChange-Id: Ie22a85dbfb9e290d432103d7d83176b0ab067fc7\nSigned-off-by: Jon Bernard \u003cjobernar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/412a3e6b6f518fda34b3ebe71aa9a5f2e173e655"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/412a3e6b6f518fda34b3ebe71aa9a5f2e173e655"}]},"branch":"refs/heads/master"},"125ac1db1350b02012c9ffa74fe840fb38f7ec8d":{"kind":"REWORK","_number":2,"created":"2026-02-18 20:59:04.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/71/977271/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/71/977271/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/71/977271/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/71/977271/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/71/977271/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/71/977271/2"}}},"commit":{"parents":[{"commit":"168bb49fc803b9c90222150013449b3a20c63108","subject":"Merge \"Adding Rafael Weingärtner candidacy for CloudKitty\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/168bb49fc803b9c90222150013449b3a20c63108"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2026-02-18 15:26:22.000000000","tz":-300},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2026-02-18 20:58:51.000000000","tz":-300},"subject":"Add Jon Bernard candidacy for Cinder PTL","message":"Add Jon Bernard candidacy for Cinder PTL\n\nChange-Id: Ie22a85dbfb9e290d432103d7d83176b0ab067fc7\nSigned-off-by: Jon Bernard \u003cjobernar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/125ac1db1350b02012c9ffa74fe840fb38f7ec8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/125ac1db1350b02012c9ffa74fe840fb38f7ec8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
