)]}'
{"id":"openstack%2Felection~977273","triplet_id":"openstack%2Felection~master~Ib229bb8d9b21fa6f45b949cd69f51cee401cb742","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-02-18 21:01:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib229bb8d9b21fa6f45b949cd69f51cee401cb742","subject":"Add Ghanshyam candidacy for QA PTL","status":"MERGED","created":"2026-02-18 20:29:21.000000000","updated":"2026-02-18 21:02:17.000000000","submitted":"2026-02-18 21:01:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977273","meta_rev_id":"e39de24ef3fe1d612dc7a0de2e2aeab964e0eade","_number":977273,"virtual_id_number":977273,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-18 21:01:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2026-02-18 20:52:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2026-02-18 20:52:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-18 20:39:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-18 20:52:04.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"508e2edbb5b220275d5bc3aea404ea081e101900","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-18 20:29:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f5180718b7d78b3befc917136d0f15d6f544d5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 20:39:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ef0bf5772854f1b9308b1f109cc5507\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9bcb37de9db49bf99580f87036a2861 : SUCCESS in 2m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/857c7f9834ae4100b3be533ddf07944c : SUCCESS in 1m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c6f2052a2404ac0af036fa8cc3aa7bd : SUCCESS in 1m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2538519f3f641718c850855e2bba0b6 : SUCCESS in 1m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3f6e6f1e895f41c8a83cf7daff052d8d : SUCCESS in 3m 21s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/925da3632b75431ea33b7304059270b4 : SUCCESS in 4m 52s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/18c7d4d28d9741bf80bae8805603bd1b : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":1},{"id":"06eaad9a38ffef38f11792189ae3bca7ba0b3e3f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-02-18 20:52:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c5b4b03d304e3820549b8bc974c28824b31d105","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 20:52:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c2fd1edebb9babc81a4427091ed16df75ab4c24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 21:01:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/783a74027c9b43d09989ae63f619e1fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4df7b278abae4ee2b382a338f714f487 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04ea733b096f4c06907aea38f8724bcc : SUCCESS in 1m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffc80bf3085646e28575d90a4148685c : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fba331a4544f485fb5ac5dba965ec96b : SUCCESS in 2m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f673deefa5834c9da66e05a295cf0e9f : SUCCESS in 3m 20s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/aef5de50159c4b41918326f355e4148b : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"2dab1b545e6d776f455174bb72650ae9715f9500","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 21:01:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e39de24ef3fe1d612dc7a0de2e2aeab964e0eade","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 21:02:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f968009d177417492a7329f15514c59\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/a7fae39014ec4245950848f4a1a37894 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f7328ef7855e99d2cf5479b902bbf86b976d4eab","revisions":{"f7328ef7855e99d2cf5479b902bbf86b976d4eab":{"kind":"REWORK","_number":1,"created":"2026-02-18 20:29:21.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/73/977273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/73/977273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/73/977273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/73/977273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/73/977273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/73/977273/1"}}},"commit":{"parents":[{"commit":"168bb49fc803b9c90222150013449b3a20c63108","subject":"Merge \"Adding Rafael Weingärtner candidacy for CloudKitty\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/168bb49fc803b9c90222150013449b3a20c63108"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-02-18 20:25:21.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-02-18 20:29:15.000000000","tz":0},"subject":"Add Ghanshyam candidacy for QA PTL","message":"Add Ghanshyam candidacy for QA PTL\n\nChange-Id: Ib229bb8d9b21fa6f45b949cd69f51cee401cb742\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/f7328ef7855e99d2cf5479b902bbf86b976d4eab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/f7328ef7855e99d2cf5479b902bbf86b976d4eab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
