)]}'
{"id":"openstack%2Felection~981312","triplet_id":"openstack%2Felection~master~I13582033d8b82039aeedd4bfaa99bad244cc2a3d","project":"openstack/election","branch":"master","topic":"2026.2-elections","attention_set":{},"removed_from_attention_set":{"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2026-03-20 17:32:06.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-03-20 17:32:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I13582033d8b82039aeedd4bfaa99bad244cc2a3d","subject":"Update tc-election-summary.py: 2026.2 results as comment","status":"MERGED","created":"2026-03-19 15:42:56.000000000","updated":"2026-03-20 17:33:05.000000000","submitted":"2026-03-20 17:32:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981312-2026.2-elections","meta_rev_id":"9602be591e721ac737e86f25012f236d43eae20a","_number":981312,"virtual_id_number":981312,"owner":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-20 17:32:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2026-03-20 14:35:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2026-03-20 14:35:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-03-20 17:08:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2026-03-20 17:08:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-19 15:51:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-20 14:35:22.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-03-20 14:35:59.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-03-20 17:08:18.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"}],"messages":[{"id":"3a394d5d87c96c24bbc9e535e55d7095b477172b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2026-03-19 15:42:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ffed342b40b7c8621df15f0c541a87e99de8816","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 15:51:31.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13937a2f516f44b9af97df8d33e4c166\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b136a65d186c4ab89fa1629eb0822ad4 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79fe3c2e047f40b5a0a43fe233c95281 : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e2859c4049345bca54e22eef8ee1917 : SUCCESS in 3m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bb6a38dd92bb4bcbbe08c8e99d542d66 : FAILURE in 5m 14s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/a96dabb3ccbb401487c4c5249f64187b : SUCCESS in 5m 30s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/5ea078f49c014cfa9e29a06c6f15bc64 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"149768306d13df19b2eeb74cccf78fec385433af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-03-20 14:35:18.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bd4c8e1d9e362f6560d6d7a2d9216851ba05f3f1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-03-20 14:35:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"65256e8a56fbbed0694c8f2342abc1f5c71e5ffe","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-20 14:35:59.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"196f4bef93686c538b03723b68077835a939ec11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 14:45:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfbc4302ebde4ecc9f2f43a8475ec51a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94b53318680b49d093c772b7d38d2fde : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df2d9524bf4b49bc8f08a72d61781354 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ef58c9db0daf4b2aba1fe5e7dec41c58 : SUCCESS in 4m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/953dce10b7ef4c1ab24a0ab4eedd9f8c : SUCCESS in 3m 20s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/091b156eb3e14a27a17502856ea7b001 : SUCCESS in 3m 25s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/441d005e02b14790912e025920c2c41b : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":2},{"id":"4e8a771ab4d7c3638fa751ea8fb25fe723889012","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2026-03-20 17:08:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3b17a5397f4e9574887e5b48c35cb4750e712fee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 17:08:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5860260b5c6f7aff96f70b3eb0dd8e6be7123dd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 17:32:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b06a481f797e4a23ae2a51fd261eb828\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e8a8fe528eb4dfd86fe4fb0fc4e294a : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/768aa52fd9b74cdf8332da93d913326c : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0037e5d1837b4ba1b7332345aaed4c4d : SUCCESS in 3m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/eb8f7f51d75b4c69969e81fa208d9412 : SUCCESS in 2m 06s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/95f4b642dffb43e6a5f7bb71821944e1 : SUCCESS in 2m 04s","accounts_in_message":[],"_revision_number":2},{"id":"5a299cafc89e87ff8464328d62a157afb1a828df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 17:32:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9602be591e721ac737e86f25012f236d43eae20a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 17:33:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee5ce1b20766493fac5e7612e4682bea\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/8148ecf112a740cdaca4dc66899de852 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3a5d56f771a15a0838b3dcba34b9d146ec655148","revisions":{"862a4f84e03e766c1619ec56461aa8df90a81dd8":{"kind":"REWORK","_number":1,"created":"2026-03-19 15:42:56.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/12/981312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/12/981312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/12/981312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/12/981312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/12/981312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/12/981312/1"}}},"commit":{"parents":[{"commit":"290bbda6d5a128859a3ec5a5a81085ab15890bc4","subject":"Close 2026.2 Election Results (TC)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/290bbda6d5a128859a3ec5a5a81085ab15890bc4"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2026-03-19 15:41:48.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2026-03-19 15:41:48.000000000","tz":540},"subject":"Update tc-election-summary.py: 2026.2 results as comment","message":"Update tc-election-summary.py: 2026.2 results as comment\n\nReflect 2026.2 election results to tc-election-summary election tool\nby mentioning that poll did not happen as comments.\n\nChange-Id: I13582033d8b82039aeedd4bfaa99bad244cc2a3d\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/862a4f84e03e766c1619ec56461aa8df90a81dd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/862a4f84e03e766c1619ec56461aa8df90a81dd8"}]},"branch":"refs/heads/master"},"3a5d56f771a15a0838b3dcba34b9d146ec655148":{"kind":"REWORK","_number":2,"created":"2026-03-20 14:35:18.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/12/981312/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/12/981312/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/12/981312/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/12/981312/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/12/981312/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/12/981312/2"}}},"commit":{"parents":[{"commit":"290bbda6d5a128859a3ec5a5a81085ab15890bc4","subject":"Close 2026.2 Election Results (TC)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/290bbda6d5a128859a3ec5a5a81085ab15890bc4"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2026-03-19 15:41:48.000000000","tz":540},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-03-20 14:35:18.000000000","tz":0},"subject":"Update tc-election-summary.py: 2026.2 results as comment","message":"Update tc-election-summary.py: 2026.2 results as comment\n\nReflect 2026.2 election results to tc-election-summary election tool\nby mentioning that poll did not happen as comments.\n\nChange-Id: I13582033d8b82039aeedd4bfaa99bad244cc2a3d\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3a5d56f771a15a0838b3dcba34b9d146ec655148"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3a5d56f771a15a0838b3dcba34b9d146ec655148"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
