)]}'
{"id":"openstack%2Fetcd3gw~1004451","triplet_id":"openstack%2Fetcd3gw~master~I47969858a4cfb39d4db58f1b8e0a6aa76eda5084","project":"openstack/etcd3gw","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-07 21:57:17.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-07 23:08:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I47969858a4cfb39d4db58f1b8e0a6aa76eda5084","subject":"Resolve setuptools deprecation warnings","status":"MERGED","created":"2026-09-07 16:05:47.000000000","updated":"2026-09-07 23:09:27.000000000","submitted":"2026-09-07 23:08:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004451","meta_rev_id":"1f9f9c4bd0f77e6f7d1f0b3d272b2651a33c95e6","_number":1004451,"virtual_id_number":1004451,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-07 23:08:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-07 21:57:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-09-07 21:55:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-07 21:57:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 17:31:54.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2026-09-07 17:32:05.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-09-07 19:05:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-07 21:55:01.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"d6c392b5a3b311935243101dd2e0a4629a7eb31d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-07 16:05:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2472a9d95f7c2ceead6503126e5dd2c209e8d968","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 17:31:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a5e9fd62b041e3fa957f235b1d48aa4ba9a0307","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 17:32:05.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"40ddc8e38ef14afc9a1ebb4b4303cdf6bea53a7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-07 17:34:48.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f3dde3525cbeb8225f2aef823626cf44273479a3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-07 17:35:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"52d6e31bc6f17c8b046f88b2449689155e177e5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 19:05:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1748be1b5ac24073a285acbd7d6ab175\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3245724627554ac6a6dbccb1b08a007b : SUCCESS in 2m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9a02bb6a55a143d29270ef4ca0ca81db : SUCCESS in 2m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28d7b82b7c364433b89fb38fa13c60ab : SUCCESS in 1h 25m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5220a701b0f4f1983b267f0d6eaa655 : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a6e25ccc71d545ac8c2115566de79783 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/204297c0043f41f094cdad009b58a464 : SUCCESS in 5m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a8584c7cdb5843cabbe34ec4b059f5e9 : SUCCESS in 5m 49s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/5f9d34f906184cbab2704c8915fa0522 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a5ab8863b9b4a24856dc5023bb0fb05 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed5669e1bb0647aba82fd17ba17e8ba9 : SUCCESS in 3m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bf6c7dfd19b4fc78fffcd0acf8d3e12 : SUCCESS in 3m 34s\n- etcd3gw-py312-functional https://zuul.opendev.org/t/openstack/build/edcca160e0c34796a3acb6f9c212ed2c : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":2},{"id":"cee52e7934ea9d3b77778453672accf3fb3e8f2a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-09-07 21:55:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8d33bae9401e82155f490b2b2130bb0f990b665f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 21:57:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cd1dd16ca22d5930becac36c4e0c5af9adc7668c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 21:57:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"96380da9d917efcfc9f6dd1d6507fa8dd9860877","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 23:08:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61f86bd9d19443149ba22e0fee1a7928\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9e3df93b6fcf472f8c7e88cb36a4d48d : SUCCESS in 2m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b2c77e32c42943f69d259617a465a3d3 : SUCCESS in 2m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7cc9e95ca4164bb3aecad7d123c2661e : SUCCESS in 1h 07m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0df3b32fef74d23b2745ba7b89dcb4e : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/92f4215a032f4109beaf9e1123735926 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/36f8f1cf4a7f4f20b63cba5475399ecd : SUCCESS in 3m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5407e0f8a961403bb6e1760474c5e520 : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27b819339f444e1282a1fe4cb64eb3e2 : SUCCESS in 3m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6538e992158a4913adefd23b6c9ac8f1 : SUCCESS in 3m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/964d6c216019420888394c5f88dc1824 : SUCCESS in 4m 24s\n- etcd3gw-py312-functional https://zuul.opendev.org/t/openstack/build/9637372530154c26928e3409b90989a3 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":2},{"id":"f3ffbde293a3e84089d53095884097308776adab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 23:08:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1f9f9c4bd0f77e6f7d1f0b3d272b2651a33c95e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 23:09:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/961c9bfb8dbc40e1aa7018f48e763ec0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c136c8aa39414232a9c80404782adc0c : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e19963c3998d3fae39e6a9c4bfc7feabd825111d","revisions":{"1ee193f2a8ecddc35587c3093bd6d845429ceaa6":{"kind":"REWORK","_number":1,"created":"2026-09-07 16:05:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/1004451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/51/1004451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/51/1004451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/51/1004451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/51/1004451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/51/1004451/1"}}},"commit":{"parents":[{"commit":"a6f3ebebbdf719fba73ec02edb9fe5054889618f","subject":"Merge \"Use etcd from distribution\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/a6f3ebebbdf719fba73ec02edb9fe5054889618f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-07 16:05:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-07 16:05:25.000000000","tz":540},"subject":"Resolve setuptools deprecation warnings","message":"Resolve setuptools deprecation warnings\n\nChange-Id: I47969858a4cfb39d4db58f1b8e0a6aa76eda5084\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/1ee193f2a8ecddc35587c3093bd6d845429ceaa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/1ee193f2a8ecddc35587c3093bd6d845429ceaa6"}]},"branch":"refs/heads/master"},"e19963c3998d3fae39e6a9c4bfc7feabd825111d":{"kind":"REWORK","_number":2,"created":"2026-09-07 17:34:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/1004451/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/51/1004451/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/51/1004451/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/51/1004451/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/51/1004451/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/51/1004451/2"}}},"commit":{"parents":[{"commit":"a6f3ebebbdf719fba73ec02edb9fe5054889618f","subject":"Merge \"Use etcd from distribution\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/a6f3ebebbdf719fba73ec02edb9fe5054889618f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-07 16:05:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-07 17:34:40.000000000","tz":540},"subject":"Resolve setuptools deprecation warnings","message":"Resolve setuptools deprecation warnings\n\nChange-Id: I47969858a4cfb39d4db58f1b8e0a6aa76eda5084\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/e19963c3998d3fae39e6a9c4bfc7feabd825111d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/e19963c3998d3fae39e6a9c4bfc7feabd825111d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
