)]}'
{"id":"openstack%2Fetcd3gw~747450","triplet_id":"openstack%2Fetcd3gw~master~Ic375cda9ec40083976e18a915db7fe8e53f47c73","project":"openstack/etcd3gw","branch":"master","topic":"fix-gitreview","hashtags":[],"change_id":"Ic375cda9ec40083976e18a915db7fe8e53f47c73","subject":"[CI] Add jobs","status":"MERGED","created":"2020-08-21 17:28:08.000000000","updated":"2021-01-03 09:22:57.000000000","submitted":"2021-01-03 09:19:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747450-fix-gitreview","meta_rev_id":"6e0e43b8404cc22d7bbdf55747fd14ffb685da02","_number":747450,"virtual_id_number":747450,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-03 09:19:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-01-03 05:05:11.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2021-01-03 09:11:17.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2021-01-03 09:11:17.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-19 15:30:11.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-25 08:40:09.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-10-25 14:47:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-03 05:05:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"e9ef961e831f344077a4ab512f7928f1b84a1479","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-21 17:28:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e863c3c32dad0ad610f4768d21aaa534ff2951da","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-21 17:28:52.000000000","message":"Patch Set 1:\n\nthrowing a very complete set of jobs at it, will likely fail at something :-)","accounts_in_message":[],"_revision_number":1},{"id":"de5425b76e799b0495b1bd6eca279350eb99a274","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 19:06:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5ed33a3682664f0ba46116832ff0f7ba : FAILURE in 2m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eff22def5f684181b5cea8506c47a224 : SUCCESS in 1h 25m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec2cd40f61c644bfbfc25e71465e3459 : FAILURE in 2m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b498cf3f2b74679b448800ca53cb290 : FAILURE in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/decda5e15dc54d008fb80d08ff7e8ac5 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb170dfcff274b49bebafc06a51f2830 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd6b34fbd2374e609e458d3fc14e1bf0 : SUCCESS in 8m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37fc7acec3de4702bf73384021166cd1 : SUCCESS in 3m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb29cd1c1d5f42e3aaa99f249e671e3f : FAILURE in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"818a110ccff6d93fd11908afa681f1a3de5cfcca","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-26 16:36:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bafaf10e372d904e767abeac962a6d424a8a102d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-26 17:45:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e1e44844f52e4b0da21630afc2a99962 : FAILURE in 1m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c2cbf7b83214b809e8a9afc0da5d909 : FAILURE in 1h 08m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/336ebf90a363466997a260ba0904ff56 : FAILURE in 2m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa3ccfacf0b64cbe8f7af0d0f4d4b28e : FAILURE in 2m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bd334b438464eee8663f9c6b204de22 : SUCCESS in 2m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d97f08f3122c4f4fa19476dc35d5e096 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc09559b8b7345258344fbc5f39bf23e : SUCCESS in 2m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54eead58098a4c6ab5608fe9e6ce0a72 : SUCCESS in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d058c1b36434ae9a9ef8a48c010e11b : FAILURE in 2m 48s","accounts_in_message":[],"_revision_number":1},{"id":"2cf0fc51b6a6ef1d456c0651eabb39fbc8c396e1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-19 15:30:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a0c2e3d3d07ea775545ff2babf4adbede25d2d17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 17:00:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d17f8a37f25b46189c5b77fee477a61a : FAILURE in 3m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f944121ac7fc4c62a2caa957c9b81bac : SUCCESS in 1h 20m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d83590071574e74ae11cf918743c122 : FAILURE in 4m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bcf104e32b5f43b39e2e1350bb3c2330 : FAILURE in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55692361c9c04ed89228ef3bfd7228ce : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07f6867305894de6beec62afeefbb686 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0ca7c1739024fafb6c0bc7979b66083 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0ea5aeb28ac4997addae2d3e704ab0c : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae7fa0756c6743e0bf41e54e848a1a13 : FAILURE in 5m 47s","accounts_in_message":[],"_revision_number":1},{"id":"465b34c6fc1bed33684eaad83da5817ff32634a2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-25 08:38:47.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7ea2560a5f802c908bbdf5a36419305cae754dc1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-25 08:39:31.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"846a36a9068c9f4d81333e6d6eda8601b5e2ffad","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-25 08:40:35.000000000","message":"Patch Set 3:\n\nlet\u0027s merge with fixmes for now to get a working gate for etcd3gw","accounts_in_message":[],"_revision_number":3},{"id":"b88dce6632a989b5ead2a9327770c27a46644fd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-25 08:48:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fa3a7b40a1946eaabdf12827485859c : SUCCESS in 2m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eaa4cf692d3741e28c2bfdc03a24fee2 : SUCCESS in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7a65bd7c4bf44fdaea70404949bfe5b : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70506e43fabe417f9860dc9bd8a41ed5 : SUCCESS in 2m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82ad687fd745448ea09f30d27fd17085 : FAILURE in 2m 35s","accounts_in_message":[],"_revision_number":3},{"id":"e851d0249a5a08b9d3ac84c7eefe3142dcaec0e0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-25 14:31:19.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"01d3b9de036a9b980eee50bc0077de46ab5e18c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-25 14:37:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30a91e5b6a23488da6343b99e30767e9 : SUCCESS in 2m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b2faaef6bab44afbe92fa3b6fa4c330 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48ca5b275cc741e08dd620963ac21a83 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94ab9a972c0f42dc883c797e8c819a9c : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":4},{"id":"9650a6cf983e5cab0042bd6aed50bf07f0da5842","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-25 14:39:06.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"9944633a47085ef3c1cb009569bf04de570306e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-25 14:47:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d9cccb94fb74333935b604e938f3d32 : SUCCESS in 2m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a56f8cd86b0440ca4227b45f516bb73 : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ca68f6d88e44dada590ebb942f280d8 : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbcb2ea029d04908b24850b82fa3342c : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":5},{"id":"cb1b99b5ec7b189762931f8dab812d28d1eddec1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-01-03 05:05:11.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"0a186799ec00c695c15c3c4debbfc87ff9ad45c9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-03 09:11:17.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nLet\u0027s get this repo mergin\u0027!","accounts_in_message":[],"_revision_number":5},{"id":"f4ef0473adf8684c20e9f535c1b57df702f86d32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 09:11:33.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e638c49a560802ce975f54cf8e59aa17373c9bc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 09:19:26.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3365b841fd514dcf90eb31859df5fe78 : SUCCESS in 2m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbcb9150e60e4f06a7ae9d2796c2d068 : SUCCESS in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52974462f64b42c39de7058e04caf78f : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/342892d99ba74219b5e56d63f49dbe83 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":5},{"id":"859ebcfcc3dd771275b1f9fd5a5310e4be9d1409","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 09:19:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"6e0e43b8404cc22d7bbdf55747fd14ffb685da02","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 09:22:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d776ff3445dc4f9bb197b304e3cce665 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"56b6839022f6672b7f4025a43da0c97035fa032e","revisions":{"4cf6349335cfabd7cf80f8053b343b774b5bad19":{"kind":"REWORK","_number":1,"created":"2020-08-21 17:28:08.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/50/747450/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/50/747450/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/50/747450/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/50/747450/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/50/747450/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/50/747450/1"}}},"commit":{"parents":[{"commit":"32004112b16960ab823ffaf740b6d354c70252e5","subject":"Fix .gitreview file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/32004112b16960ab823ffaf740b6d354c70252e5"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-21 17:27:30.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-21 17:27:30.000000000","tz":120},"subject":"[CI] Add jobs","message":"[CI] Add jobs\n\nChange-Id: Ic375cda9ec40083976e18a915db7fe8e53f47c73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/4cf6349335cfabd7cf80f8053b343b774b5bad19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/4cf6349335cfabd7cf80f8053b343b774b5bad19"}]},"branch":"refs/heads/master"},"d783069a144de1e51dc53f919c0f15d2dac9eccd":{"kind":"REWORK","_number":2,"created":"2020-10-25 08:38:47.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/50/747450/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/50/747450/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/50/747450/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/50/747450/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/50/747450/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/50/747450/2"}}},"commit":{"parents":[{"commit":"32004112b16960ab823ffaf740b6d354c70252e5","subject":"Fix .gitreview file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/32004112b16960ab823ffaf740b6d354c70252e5"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-21 17:27:30.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-10-25 08:38:13.000000000","tz":0},"subject":"[CI] Add jobs","message":"[CI] Add jobs\n\nChange-Id: Ic375cda9ec40083976e18a915db7fe8e53f47c73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/d783069a144de1e51dc53f919c0f15d2dac9eccd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/d783069a144de1e51dc53f919c0f15d2dac9eccd"}]},"branch":"refs/heads/master"},"de28aafe3fa241a95b5ebac6cd94bd7788b26029":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-10-25 08:39:31.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/50/747450/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/50/747450/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/50/747450/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/50/747450/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/50/747450/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/50/747450/3"}}},"commit":{"parents":[{"commit":"7a1a2b5a672605ae549c73ed18302b7abd9e0e30","subject":"Restore previous endpoint default (v3alpha)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/7a1a2b5a672605ae549c73ed18302b7abd9e0e30"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-21 17:27:30.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-10-25 08:39:31.000000000","tz":0},"subject":"[CI] Add jobs","message":"[CI] Add jobs\n\nChange-Id: Ic375cda9ec40083976e18a915db7fe8e53f47c73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/de28aafe3fa241a95b5ebac6cd94bd7788b26029"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/de28aafe3fa241a95b5ebac6cd94bd7788b26029"}]},"branch":"refs/heads/master"},"cbf203d997d52a0d0e08ce793fe864707d518b9b":{"kind":"REWORK","_number":4,"created":"2020-10-25 14:31:19.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/50/747450/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/50/747450/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/50/747450/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/50/747450/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/50/747450/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/50/747450/4"}}},"commit":{"parents":[{"commit":"7a1a2b5a672605ae549c73ed18302b7abd9e0e30","subject":"Restore previous endpoint default (v3alpha)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/7a1a2b5a672605ae549c73ed18302b7abd9e0e30"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-21 17:27:30.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-10-25 14:31:15.000000000","tz":0},"subject":"[CI] Add jobs","message":"[CI] Add jobs\n\nChange-Id: Ic375cda9ec40083976e18a915db7fe8e53f47c73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/cbf203d997d52a0d0e08ce793fe864707d518b9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/cbf203d997d52a0d0e08ce793fe864707d518b9b"}]},"branch":"refs/heads/master"},"56b6839022f6672b7f4025a43da0c97035fa032e":{"kind":"REWORK","_number":5,"created":"2020-10-25 14:39:06.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/50/747450/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/50/747450/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/50/747450/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/50/747450/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/50/747450/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/50/747450/5"}}},"commit":{"parents":[{"commit":"7a1a2b5a672605ae549c73ed18302b7abd9e0e30","subject":"Restore previous endpoint default (v3alpha)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/7a1a2b5a672605ae549c73ed18302b7abd9e0e30"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-21 17:27:30.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-10-25 14:39:03.000000000","tz":0},"subject":"[CI] Add jobs","message":"[CI] Add jobs\n\nChange-Id: Ic375cda9ec40083976e18a915db7fe8e53f47c73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/56b6839022f6672b7f4025a43da0c97035fa032e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/56b6839022f6672b7f4025a43da0c97035fa032e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
