)]}'
{"id":"openstack%2Fetcd3gw~836244","triplet_id":"openstack%2Fetcd3gw~master~I6319d58462481aac4b15048aad57de77094e8769","project":"openstack/etcd3gw","branch":"master","topic":"bug/1967674","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-07-05 14:50:10.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"33910":{"account":{"_account_id":33910,"name":"Tyler Stachecki","display_name":"Tyler Stachecki","email":"tstachecki@bloomberg.net","username":"tjstachecki"},"last_update":"2022-04-08 06:25:59.000000000","reason":"Reviewer/Cc was removed"},"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2022-04-06 14:44:08.000000000","reason":"Removed by Ekaterina Chernova by clicking the attention icon"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-04-14 14:04:56.000000000","reason":"Hervé Beraud replied on the change"},"7549":{"account":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"last_update":"2022-07-05 14:55:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6319d58462481aac4b15048aad57de77094e8769","subject":"Handle key has only metadata and no value","status":"MERGED","created":"2022-04-03 14:39:41.000000000","updated":"2022-07-05 14:56:35.000000000","submitted":"2022-07-05 14:55:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836244-bug/1967674","meta_rev_id":"0dbc778122fa5a165af1b07d021ff59db7363118","_number":836244,"virtual_id_number":836244,"owner":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-05 14:55:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-07-05 14:50:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2022-04-14 14:04:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-07-05 14:50:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"}],"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-03 14:48:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-04 08:49:47.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"CC"},{"updated":"2022-04-05 13:25:54.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2022-04-06 14:44:39.000000000","updated_by":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-04-06 14:44:39.000000000","updated_by":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"reviewer":{"_account_id":33910,"name":"Tyler Stachecki","display_name":"Tyler Stachecki","email":"tstachecki@bloomberg.net","username":"tjstachecki"},"state":"REVIEWER"},{"updated":"2022-04-08 06:25:59.000000000","updated_by":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"reviewer":{"_account_id":33910,"name":"Tyler Stachecki","display_name":"Tyler Stachecki","email":"tstachecki@bloomberg.net","username":"tjstachecki"},"state":"REMOVED"},{"updated":"2022-04-14 14:04:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"55cd9428b1148395dffed2ae3eec88383c24065b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"date":"2022-04-03 14:39:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6811de46213c28aeaffc95e312d2aeff83e092e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-03 14:48:29.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/703a74b6801e44a6ae6e9a6a0ec312f5 : POST_FAILURE in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/270ea37d13024cefab11374e939586f7 : FAILURE in 2m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/963d1a942196412d88113f3285aaf6c2 : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/374720db41ab4db5b0cf871f4e1b9c90 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed30dec55a0542febc6de27be5c7d31c : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29fa58b6fc4649c684da0557bd4c55ba : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"353c39dec9531fd051b3e26dab9e63bd4ae7abb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"date":"2022-04-04 07:39:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3932cd481440529c9fa9639efbb9ebd503754ae5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 07:45:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/207735df89264304a7ad0576f6ddeea7 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8392c22e36d348b1b904c2ba86b2a927 : SUCCESS in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e8d35e06d9549cf80da3b163242a43d : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0e93900ad6244dda6a8ce8dd4b7c533 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba1896920d2f49c0a5a44656e7e63cf2 : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e6cf0ece61a4bceb5193bc7ab05ec2f : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":2},{"id":"8d35dc6c55933fac43d398f1fb9b9a74d913a25a","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2022-04-04 08:49:47.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"349d388f6423d93b5b8932488f7bdc07c77a6474","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"date":"2022-04-05 13:09:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7a852bd7ecc4cd68b7c3d90e7d7e0ce3a854066f","author":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"date":"2022-04-05 13:10:27.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b21dd7e3f5db97a68945c4e208fc706e65947cc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"date":"2022-04-05 13:10:50.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"38666358a2d48621bbd9c86212b5415fa4755adf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 13:24:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5946bcbe5ac446ab8d659783b5d738a4 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1b3b086f3f04649b3ef5aae67e4b8fd : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92b19ba9975041e1903ef52ba55411df : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e06b59c73574998a6d3fcaae61fd84c : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c528a031978444b9c7618044380cc86 : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b75534bfeab4925821fa79e62d4c8fe : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":4},{"id":"3ddf33f94e1a633b89b1f58676d3e10a776c3c41","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-05 13:25:54.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cea71872cb6380e6d7b4350466f52cadfc280229","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"date":"2022-04-06 13:26:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0d3f0acff4aff49d7f9c8e01bab30ea9fbf0b34b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 13:38:17.000000000","message":"Patch Set 5: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f56abd82aa74f1eaf15f19174c5757d : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9259a1e8bc640ac9cc9d0920faa4f20 : FAILURE in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c4ea96a0d39484f8e0213ce4f545653 : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce5216812d8f4ca9865c128f474c631c : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba523e4d698148cc8225795bfde2e30f : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bb449c7783748fd84ba30545e17d7ef : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":5},{"id":"3ac640ed89bc880d7fa828a434df9ab35bd38cd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"date":"2022-04-06 13:53:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5fa41531c0ea387b3b52ebc87b7eade33b4807c0","author":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"date":"2022-04-06 13:57:19.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"120004810a6e6a3b3e3e19f5938d8a9020374bac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 13:59:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da5f8dded6204dd8ae3a530e345c6c27 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8a5d5cc8cd64892886199fb33b12765 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a180b45ba164f7da4985f3b032c0ebc : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd9aa6e6669c47dbaa61dfddfc5e5e2b : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85cc25dd14254a2ab45bb3bc6a968af3 : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7e7b0cad9134bf4ac6f25bcb70eb7d2 : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72ffe1ef52be42ee9e55343121bc3043 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":6},{"id":"e54d9a3d11e4278fe7d95807a1b3d459c7ada1aa","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"date":"2022-04-08 06:25:59.000000000","message":"Removed reviewer Tyler Stachecki.","accounts_in_message":[],"_revision_number":6},{"id":"b92023cdf2004cf841e601aa5bcd7fcb635fb090","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"date":"2022-04-13 06:11:04.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"8aa3f7b98fe1fad095739e79bc74b36ffd869445","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 06:18:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6271f1651c4d4dd490e15c6bac698231 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a113b7bbd95423bb552e50e475f8d1a : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7154f252aaff4d1f9d5615215ca04f0e : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/073b7de4ac1e43588e42ec6469bdb5d6 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a81ccf7077044998812ab4a51d4d070 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a0b3c4f385942d4b888c9db9cc0e95c : SUCCESS in 3m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24897b830b934656a39c0017151188e3 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":7},{"id":"47f9054d546c063c126f5f57d5393ff92d86bb03","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-14 14:04:56.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b1994314da2b1f42bf0f7a86dcb23f51ee00cb12","author":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"date":"2022-04-15 08:44:18.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0c47f1ecf931fe10524e6947ed9eaf47de29847c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-05 14:50:10.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"ecd23cdb3d43652233abf729a024ae83aebaf96e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 14:50:37.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"e61a455302ddb8070dcf2230c14a97b166ba1485","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 14:55:09.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8319d471ae524d969ee585c55ae637bb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63bd1a48abb64d149623dee55f827234 : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b35042f01f2a4190bd38b9f71befd76a : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2eb6d294e5a4b96ab113373d0068e34 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/649b057867384b98b928e4ac8e7e05a9 : SUCCESS in 3m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0258be469da49068531aca6bb173f28 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":7},{"id":"a119de230a56753fb7359c44666ac1ccb8ad2448","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 14:55:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"0dbc778122fa5a165af1b07d021ff59db7363118","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 14:56:35.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26f3ca3832fd453e949ac2c9ad96ce4d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f44c072bfa8c4fc69e00d9c71ec82b04 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a321080601e4ac99c47e956efc5081b : SUCCESS in 52s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"e00c9f2ec4205650aff7ffd8972591b9f8c45ae0","revisions":{"aa1f79ca0938603911c4bde393b3e4ec12531c40":{"kind":"REWORK","_number":1,"created":"2022-04-03 14:39:41.000000000","uploader":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"ref":"refs/changes/44/836244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/44/836244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/1"}}},"commit":{"parents":[{"commit":"9b107f085b40b0d6bf83ef20027907aaef8ef4b9","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/9b107f085b40b0d6bf83ef20027907aaef8ef4b9"}]}],"author":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-03 12:55:48.000000000","tz":180},"committer":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-03 14:39:12.000000000","tz":180},"subject":"Handle key has only metadata and no value","message":"Handle key has only metadata and no value\n\nIf we have such response without \u0027value\u0027\n {\u0027key\u0027: \u0027dG9vei9ncm91cHMvYWxhcm1fZXZhbHVhdG9yLw\u003d\u003d\u0027, \u0027create_revision\u0027: \u00272\u0027, \u0027mod_revision\u0027: \u00272\u0027, \u0027version\u0027: \u00271\u0027}\nclient will send empty binary string instead of crashing\n\nCloses-bug: #1967674\nChange-Id: I6319d58462481aac4b15048aad57de77094e8769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/aa1f79ca0938603911c4bde393b3e4ec12531c40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/aa1f79ca0938603911c4bde393b3e4ec12531c40"}]},"branch":"refs/heads/master"},"728a2b23eaa0ba8b797d6e849cae6fe51d8d93e8":{"kind":"REWORK","_number":2,"created":"2022-04-04 07:39:40.000000000","uploader":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"ref":"refs/changes/44/836244/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/44/836244/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/2"}}},"commit":{"parents":[{"commit":"9b107f085b40b0d6bf83ef20027907aaef8ef4b9","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/9b107f085b40b0d6bf83ef20027907aaef8ef4b9"}]}],"author":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-03 12:55:48.000000000","tz":180},"committer":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-04 07:38:59.000000000","tz":180},"subject":"Handle key has only metadata and no value","message":"Handle key has only metadata and no value\n\nIf we have such response without \u0027value\u0027\n {\u0027key\u0027: \u0027dG9vei9ncm91cHMvYWxhcm1fZXZhbHVhdG9yLw\u003d\u003d\u0027, \u0027create_revision\u0027: \u00272\u0027, \u0027mod_revision\u0027: \u00272\u0027, \u0027version\u0027: \u00271\u0027}\nclient will send empty binary string instead of crashing\n\nCloses-bug: #1967674\nChange-Id: I6319d58462481aac4b15048aad57de77094e8769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/728a2b23eaa0ba8b797d6e849cae6fe51d8d93e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/728a2b23eaa0ba8b797d6e849cae6fe51d8d93e8"}]},"branch":"refs/heads/master"},"8771674c5c20273179c7b5c5775ea078c7357f13":{"kind":"REWORK","_number":3,"created":"2022-04-05 13:09:57.000000000","uploader":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"ref":"refs/changes/44/836244/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/44/836244/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/3"}}},"commit":{"parents":[{"commit":"9b107f085b40b0d6bf83ef20027907aaef8ef4b9","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/9b107f085b40b0d6bf83ef20027907aaef8ef4b9"}]}],"author":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-03 12:55:48.000000000","tz":180},"committer":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-05 12:53:47.000000000","tz":180},"subject":"Handle key has only metadata and no value","message":"Handle key has only metadata and no value\n\nSome other etcd clients may add key\n without providing \u0027value\u0027.\n\nThis patch adds compatibility, so etcd can be used\ntogether with abother clients\n\nCloses-bug: #1967674\nChange-Id: I6319d58462481aac4b15048aad57de77094e8769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/8771674c5c20273179c7b5c5775ea078c7357f13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/8771674c5c20273179c7b5c5775ea078c7357f13"}]},"branch":"refs/heads/master"},"89f5af9df45a050299358327a32df289d7255ab9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-04-05 13:10:50.000000000","uploader":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"ref":"refs/changes/44/836244/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/44/836244/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/4"}}},"commit":{"parents":[{"commit":"bbd14c466e854b63f5d1ccacaa7945b8e29f177c","subject":"Remove six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/bbd14c466e854b63f5d1ccacaa7945b8e29f177c"}]}],"author":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-03 12:55:48.000000000","tz":180},"committer":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-05 13:10:50.000000000","tz":0},"subject":"Handle key has only metadata and no value","message":"Handle key has only metadata and no value\n\nSome other etcd clients may add key\n without providing \u0027value\u0027.\n\nThis patch adds compatibility, so etcd can be used\ntogether with abother clients\n\nCloses-bug: #1967674\nChange-Id: I6319d58462481aac4b15048aad57de77094e8769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/89f5af9df45a050299358327a32df289d7255ab9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/89f5af9df45a050299358327a32df289d7255ab9"}]},"branch":"refs/heads/master","description":"Rebase"},"cb71a0de3147b13a3bbea3bbefbb5f4287f2198f":{"kind":"REWORK","_number":5,"created":"2022-04-06 13:26:36.000000000","uploader":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"ref":"refs/changes/44/836244/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/44/836244/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/5"}}},"commit":{"parents":[{"commit":"9b107f085b40b0d6bf83ef20027907aaef8ef4b9","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/9b107f085b40b0d6bf83ef20027907aaef8ef4b9"}]}],"author":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-03 12:55:48.000000000","tz":180},"committer":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-06 13:24:17.000000000","tz":180},"subject":"Handle key has only metadata and no value","message":"Handle key has only metadata and no value\n\nSome other etcd clients may add key\n without providing \u0027value\u0027.\n\nThis patch adds compatibility, so etcd can be used\ntogether with another clients\n\nCloses-bug: #1967674\nChange-Id: I6319d58462481aac4b15048aad57de77094e8769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/cb71a0de3147b13a3bbea3bbefbb5f4287f2198f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/cb71a0de3147b13a3bbea3bbefbb5f4287f2198f"}]},"branch":"refs/heads/master"},"9c722e4c3a988675f4e76d5470c7219a842cdcfa":{"kind":"REWORK","_number":6,"created":"2022-04-06 13:53:22.000000000","uploader":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"ref":"refs/changes/44/836244/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/44/836244/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/6"}}},"commit":{"parents":[{"commit":"9b107f085b40b0d6bf83ef20027907aaef8ef4b9","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/9b107f085b40b0d6bf83ef20027907aaef8ef4b9"}]}],"author":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-03 12:55:48.000000000","tz":180},"committer":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-06 13:52:50.000000000","tz":180},"subject":"Handle key has only metadata and no value","message":"Handle key has only metadata and no value\n\nSome other etcd clients may add key\n without providing \u0027value\u0027.\n\nThis patch adds compatibility, so etcd can be used\ntogether with another clients\n\nCloses-bug: #1967674\nChange-Id: I6319d58462481aac4b15048aad57de77094e8769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/9c722e4c3a988675f4e76d5470c7219a842cdcfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/9c722e4c3a988675f4e76d5470c7219a842cdcfa"}]},"branch":"refs/heads/master"},"e00c9f2ec4205650aff7ffd8972591b9f8c45ae0":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-04-13 06:11:04.000000000","uploader":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"ref":"refs/changes/44/836244/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/44/836244/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/44/836244/7"}}},"commit":{"parents":[{"commit":"bbd14c466e854b63f5d1ccacaa7945b8e29f177c","subject":"Remove six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/bbd14c466e854b63f5d1ccacaa7945b8e29f177c"}]}],"author":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-03 12:55:48.000000000","tz":180},"committer":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-13 06:11:04.000000000","tz":0},"subject":"Handle key has only metadata and no value","message":"Handle key has only metadata and no value\n\nSome other etcd clients may add key\n without providing \u0027value\u0027.\n\nThis patch adds compatibility, so etcd can be used\ntogether with another clients\n\nCloses-bug: #1967674\nChange-Id: I6319d58462481aac4b15048aad57de77094e8769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/e00c9f2ec4205650aff7ffd8972591b9f8c45ae0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/e00c9f2ec4205650aff7ffd8972591b9f8c45ae0"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
