)]}'
{"id":"openstack%2Fetcd3gw~843003","triplet_id":"openstack%2Fetcd3gw~master~I440cedb711149a5f12eb2311e78181b01666d274","project":"openstack/etcd3gw","branch":"master","topic":"bug/1895952","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-07-05 14:43:42.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"32927":{"account":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"last_update":"2022-07-05 15:01:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I440cedb711149a5f12eb2311e78181b01666d274","subject":"Handle refreshing of expired lease.","status":"MERGED","created":"2022-05-23 16:47:05.000000000","updated":"2022-07-05 15:03:13.000000000","submitted":"2022-07-05 15:01:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843003-bug/1895952","meta_rev_id":"ac521cfccf87d5b295012d77efbc8986bf3df129","_number":843003,"virtual_id_number":843003,"owner":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-05 15:01:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-07-05 14:43:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-07-05 14:51:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-23 16:57:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-05 10:03:53.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2022-07-05 10:04:33.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"2fbd2560afd4d3d67e5577714518c9e670f0ac1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-05-23 16:47:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a1602fbdb3db860ec2c52bdd5b349274d6d6ead","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 16:57:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f27d6b786d54e8bbe91fec2a3fd9fcb : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a30470d3dc5847c2be30d6568014ca80 : SUCCESS in 2m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ec2172e20df412487164f66de09120b : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9c7e38f2e3644b9ab03c46e6a36eef0 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25ab15ddae0447f3b38464f7df898e90 : SUCCESS in 2m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22c6ae54b8ba479d8b0382ac7b4203ba : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"2518156d2ac6d2d9811d929e5117fad6eee84462","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-05-24 17:04:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ac5227cc1daf1b6b1913e4f7c3b37298878916f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 17:12:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc3d0108df5c4d739aa107c9be36dc7d : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62159c59690a4aa0a8230724323a6a2b : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff572fe4c50e4748990530dbbc0296d4 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ffa85b7787742bc8091e3cdfb2a23a7 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bdb992ef252466cafc3d67860d8667e : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4742102ea77649f8bd80a5219d3dac4a : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"7d429400c33f9caf9c46ce1766021beb23a9551f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-05 10:03:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2f9fb6326bb6ef292562436a006dfbb85f6e3b12","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-05 10:04:33.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"52f969530614b690b494e30a598887bc06eb7451","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-07-05 11:05:27.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8dc13bb3429918e6f25101c67fc3ec1ab0d30f6d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-05 12:52:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"165fbc2a522f0a02e2bd66722371318e503a6b97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-07-05 13:34:11.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c2f9c7ee04cf98b07843797971895382a86dd498","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-07-05 13:34:17.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c7d00ea6774bafe734e2bc7cc9d17b02a3d94ab3","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-07-05 13:34:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9c8e63033160646825e1cc19ee3bc39e8ef94e69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 13:43:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e84034847151498c934a2d237295fb27\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0f93b81ec0a4e8291c9c3d095b15811 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/246976961965407fad99c6521221a0d0 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9f05e6cc7bb43d88a091de6d25b0118 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9e2d4e0728948979e44b64ca10999fd : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a4482cdceeb4406b24bf51f61654633 : SUCCESS in 2m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ee4c0d931cb4131b6db7d2634e9e58f : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":4},{"id":"76be959b3cfa094927da3a07a3b80a148663f89f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-05 14:43:42.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"75833228db10a5b7ed8c52d3b5eb7822a6bf0e5c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-05 14:51:03.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"323b61b21a8c526ff5fe5dcfa5edb75a2a678b82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 14:51:34.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ac4388a073fbfb89f74dda4d7d0d4e2adc4d45a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 15:01:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fea866b6d91b4ea0bc463967725d88b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ee4369130e246089e15f40b5725bb6a : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8784b9da040416181d568ab3d1548fe : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8936b61f988f400c8570672e51d23fc1 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/602f96d2f4c9423aa63bf14340cff429 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":4},{"id":"b598910f5c4bcd7d5175bf53bf16730170c733bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 15:01:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"ac521cfccf87d5b295012d77efbc8986bf3df129","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 15:03:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ef61d34495742e0a3c00efd52f8b8b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/200983eb92cf4816bf39a0e269357804 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e35c7aa1f617278c17649206f98cceb8800bda98","revisions":{"879e3935dc6bfbd48c649d1eddc83c3b5f68c64c":{"kind":"REWORK","_number":1,"created":"2022-05-23 16:47:05.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/03/843003/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/03/843003/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/03/843003/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/03/843003/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/03/843003/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/03/843003/1"}}},"commit":{"parents":[{"commit":"bbd14c466e854b63f5d1ccacaa7945b8e29f177c","subject":"Remove six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/bbd14c466e854b63f5d1ccacaa7945b8e29f177c"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-05-23 16:45:04.000000000","tz":180},"committer":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-05-23 16:45:04.000000000","tz":180},"subject":"Handle refreshing of expired lease.","message":"Handle refreshing of expired lease.\n\nRefreshing of expired lease causes non-obvious error:\nRetrying tooz.drivers.etcd3gw.Etcd3Driver.heartbeat\nin 1.0 seconds as it raised KeyError: \u0027TTL\u0027.\nThe patch handles the error.\n\nCloses-Bug: 1895952\nChange-Id: I440cedb711149a5f12eb2311e78181b01666d274\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/879e3935dc6bfbd48c649d1eddc83c3b5f68c64c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/879e3935dc6bfbd48c649d1eddc83c3b5f68c64c"}]},"branch":"refs/heads/master"},"d59b426c290a0bae5626d47c3fcb8786c64451e6":{"kind":"REWORK","_number":2,"created":"2022-05-24 17:04:04.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/03/843003/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/03/843003/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/03/843003/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/03/843003/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/03/843003/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/03/843003/2"}}},"commit":{"parents":[{"commit":"bbd14c466e854b63f5d1ccacaa7945b8e29f177c","subject":"Remove six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/bbd14c466e854b63f5d1ccacaa7945b8e29f177c"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-05-23 16:45:04.000000000","tz":180},"committer":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-05-24 16:57:25.000000000","tz":180},"subject":"Handle refreshing of expired lease.","message":"Handle refreshing of expired lease.\n\nRefreshing of expired lease causes non-obvious error:\nRetrying tooz.drivers.etcd3gw.Etcd3Driver.heartbeat\nin 1.0 seconds as it raised KeyError: \u0027TTL\u0027.\nThe patch handles the error.\n\nCloses-Bug: 1895952\nChange-Id: I440cedb711149a5f12eb2311e78181b01666d274\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/d59b426c290a0bae5626d47c3fcb8786c64451e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/d59b426c290a0bae5626d47c3fcb8786c64451e6"}]},"branch":"refs/heads/master"},"811fc6def7b3a4058ec9084cf65dd70add0486e5":{"kind":"REWORK","_number":3,"created":"2022-07-05 13:34:11.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/03/843003/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/03/843003/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/03/843003/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/03/843003/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/03/843003/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/03/843003/3"}}},"commit":{"parents":[{"commit":"bbd14c466e854b63f5d1ccacaa7945b8e29f177c","subject":"Remove six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/bbd14c466e854b63f5d1ccacaa7945b8e29f177c"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-05-23 16:45:04.000000000","tz":180},"committer":{"name":"mitya-eremeev-2","email":"mitossvyaz@mail.ru","date":"2022-07-05 13:34:03.000000000","tz":0},"subject":"Handle refreshing of expired lease.","message":"Handle refreshing of expired lease.\n\nRefreshing of expired lease causes non-obvious error:\nRetrying tooz.drivers.etcd3gw.Etcd3Driver.heartbeat\nin 1.0 seconds as it raised KeyError: \u0027TTL\u0027.\nThe patch handles the error.\n\nCloses-Bug: 1895952\nChange-Id: I440cedb711149a5f12eb2311e78181b01666d274\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/811fc6def7b3a4058ec9084cf65dd70add0486e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/811fc6def7b3a4058ec9084cf65dd70add0486e5"}]},"branch":"refs/heads/master"},"e35c7aa1f617278c17649206f98cceb8800bda98":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-07-05 13:34:17.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/03/843003/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/03/843003/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/03/843003/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/03/843003/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/03/843003/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/03/843003/4"}}},"commit":{"parents":[{"commit":"0ed09452759dc021c5c1a5aa9d460eacd5a5ae70","subject":"Drop python3.6/3.7 support in testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/0ed09452759dc021c5c1a5aa9d460eacd5a5ae70"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-05-23 16:45:04.000000000","tz":180},"committer":{"name":"mitya-eremeev-2","email":"mitossvyaz@mail.ru","date":"2022-07-05 13:34:17.000000000","tz":0},"subject":"Handle refreshing of expired lease.","message":"Handle refreshing of expired lease.\n\nRefreshing of expired lease causes non-obvious error:\nRetrying tooz.drivers.etcd3gw.Etcd3Driver.heartbeat\nin 1.0 seconds as it raised KeyError: \u0027TTL\u0027.\nThe patch handles the error.\n\nCloses-Bug: 1895952\nChange-Id: I440cedb711149a5f12eb2311e78181b01666d274\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/e35c7aa1f617278c17649206f98cceb8800bda98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/e35c7aa1f617278c17649206f98cceb8800bda98"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
