)]}'
{"id":"openstack%2Fetcd3gw~873357","triplet_id":"openstack%2Fetcd3gw~master~I05ab198c2cdec2683249caf9e591221d3934eb6e","project":"openstack/etcd3gw","branch":"master","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-02-10 13:46:15.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}},"hashtags":[],"change_id":"I05ab198c2cdec2683249caf9e591221d3934eb6e","subject":"Fix the docs job","status":"MERGED","created":"2023-02-10 09:39:13.000000000","updated":"2023-02-10 14:05:14.000000000","submitted":"2023-02-10 14:04:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"873357","meta_rev_id":"777598415e386edf1f4e47a0f85e90144e7537fb","_number":873357,"virtual_id_number":873357,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-10 14:04:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-02-10 13:23:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2023-02-10 13:46:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2023-02-10 13:46:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 09:45:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-10 13:03:33.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"7671a72eaf4566d47181d16a2989bb045a8ca4d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-10 09:39:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"309a57beb052f8069d0d33688c0ef75acc14c0b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-10 09:40:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ee3cf6de10ab1d443cd9fbdbb9db37b780a571a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 09:45:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/db917635bbf6471d8801ab414eaa3a83\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c29da749ff84019be5d59ecf227a745 : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/322aeda53d984631993c4714b2e8e00e : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2caded9cd0440ecb67a3bd5e7a0a75f : SUCCESS in 2m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2ddb04dd2b9473ea03e09326c6434e7 : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe105bf1be1f4176b403d6d5847ff7b3 : FAILURE in 2m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06dfd57883144feca84bf89e8299dd80 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":2},{"id":"776dffc2eb27387b25021e8259c9c62aa0f1d159","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-10 09:56:00.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"6b2fd960eff1125e798113624d4875ec19c3998f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-10 09:56:57.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"3bd02261abe1bdb35784e1c3486d5f3f6c58d208","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 10:08:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f90fd2e0dbcd4f92b37f24432c32a3aa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0936314af9ce414093bff04dab39a681 : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd315ae8b68a4eaca332f6c8d6a03056 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/227ef139282941db89fe80cbdf5324c1 : SUCCESS in 2m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92e84bd5704845bbae70e9959702920e : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e0209e8936d406099e216812472d7bc : FAILURE in 2m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbfc1c34d0754dca9bc06d6fe48a3d1e : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":4},{"id":"d110a8dafe9f8a3c0ee2ec15425c084b1bb43f83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-10 12:09:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ef5301db10e05ae2db36be0c95ba8151cb2b8757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 12:16:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90e4223b3f6e4d21839a2ed6f6143363\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38ff44f88f4e4a48ac26b4bdbbe6ff29 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c49ee39b1fac443eafe8c4566ee629de : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94540d7ced6240c5815d87615d2ae554 : SUCCESS in 2m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5f454b7c7124d9a9b19477b66265695 : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a86e210235d540bb98a176ed4f1913d7 : SUCCESS in 3m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f429b17d9dea404fa7089694f9ca8b7d : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":5},{"id":"8c190d4cbe33735e38c3efe726223babc03be610","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-10 13:03:33.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"603030f8493e24fd45eca99cbda8d2d4cfb0f4e5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-10 13:04:45.000000000","message":"Patch Set 5: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d678096223e4230aa88be3eea80ca7e01b1c8269","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-10 13:08:38.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5a69d22f8ab0d75614a8848f516f8b570612e7fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 13:19:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae5571f0db344097a2079d7fcd3bdb5b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/652a0f92276c449c9516bc4b3c25633d : SUCCESS in 2m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7b9b9b519e34804802f481cb078cce8 : SUCCESS in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abd15141b46041c88ccde706962d7036 : SUCCESS in 2m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea0d3f08d3124a41bb7f3f34b160ddac : SUCCESS in 2m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c51e6e3db748fab5717764d7c975a4 : SUCCESS in 3m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f91b8a59a874eff9879b67beea16b04 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":6},{"id":"fc26745101b3c09204d0ea8d2828dc139a24f35b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-10 13:23:43.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"b07cf849b16c92d42a2eb3effd14527b8d32acd2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-10 13:46:15.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"96eb520e55a1243a41864382b75fcf967ffd0c40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 13:46:36.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"6d6a0d10b8ea723aac6e9f72568931101f51ea14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 14:04:09.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6522ae52fedd4af183e83d769731bb49\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f19cdfca4464d879ecb6806e28bf74b : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ca38f0cdac0483b82211c384e648a8a : SUCCESS in 2m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c4180e7776445fb8cc712aff9f484ff : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9022ba7d5566474d8218d5323fd5645b : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c36af192f5f14df69c7ab2bec0d74cb8 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":6},{"id":"5110546ccf9011bd211aaf771412a2cca3fb1ad2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 14:04:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"777598415e386edf1f4e47a0f85e90144e7537fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 14:05:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25a86b940db741b2bad4dce504afcf23\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb4cc423221848a487357d9a65d6b160 : SUCCESS in 34s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56112b29433c4e5d951e3509306b218f : SUCCESS in 43s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"1a2ddf8a91b666afd6f030736fb1b95ecc3f9a69","revisions":{"9c186818caafe4fbe5c12749c2295164d6777f2f":{"kind":"REWORK","_number":1,"created":"2023-02-10 09:39:13.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/57/873357/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/57/873357/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/1"}}},"commit":{"parents":[{"commit":"2a7f392a80572ef69ce613fa80b1392b5558d869","subject":"Pass api_path thru etcd3gw.client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/2a7f392a80572ef69ce613fa80b1392b5558d869"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-10 09:38:59.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-10 09:39:11.000000000","tz":60},"subject":"fix the tox doc job","message":"fix the tox doc job\n\nChange-Id: I05ab198c2cdec2683249caf9e591221d3934eb6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/9c186818caafe4fbe5c12749c2295164d6777f2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/9c186818caafe4fbe5c12749c2295164d6777f2f"}]},"branch":"refs/heads/master"},"9c3a17925a91883be0966ffd9ce442d99258b8d1":{"kind":"REWORK","_number":2,"created":"2023-02-10 09:40:44.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/57/873357/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/57/873357/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/2"}}},"commit":{"parents":[{"commit":"2a7f392a80572ef69ce613fa80b1392b5558d869","subject":"Pass api_path thru etcd3gw.client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/2a7f392a80572ef69ce613fa80b1392b5558d869"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-10 09:38:59.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-10 09:40:36.000000000","tz":60},"subject":"fix the tox doc job","message":"fix the tox doc job\n\nChange-Id: I05ab198c2cdec2683249caf9e591221d3934eb6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/9c3a17925a91883be0966ffd9ce442d99258b8d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/9c3a17925a91883be0966ffd9ce442d99258b8d1"}]},"branch":"refs/heads/master"},"784cb2afbafbf51a27c7d966f6e4169aa50b94e0":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-02-10 09:56:00.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/57/873357/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/57/873357/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/3"}}},"commit":{"parents":[{"commit":"2a7f392a80572ef69ce613fa80b1392b5558d869","subject":"Pass api_path thru etcd3gw.client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/2a7f392a80572ef69ce613fa80b1392b5558d869"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-10 09:38:59.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-10 09:53:25.000000000","tz":60},"subject":"fix the doc job","message":"fix the doc job\n\nDisable installing the current package in development mode\nwith develop mode.\n\nChange-Id: I05ab198c2cdec2683249caf9e591221d3934eb6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/784cb2afbafbf51a27c7d966f6e4169aa50b94e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/784cb2afbafbf51a27c7d966f6e4169aa50b94e0"}]},"branch":"refs/heads/master"},"d6a7014cacd5c1a1e50863eeb3044e68b7967c1a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-02-10 09:56:57.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/57/873357/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/57/873357/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/4"}}},"commit":{"parents":[{"commit":"2a7f392a80572ef69ce613fa80b1392b5558d869","subject":"Pass api_path thru etcd3gw.client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/2a7f392a80572ef69ce613fa80b1392b5558d869"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-10 09:38:59.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-10 09:56:43.000000000","tz":60},"subject":"Fix the docs job","message":"Fix the docs job\n\nDisable installing the current package in development mode\nwith develop mode.\n\nChange-Id: I05ab198c2cdec2683249caf9e591221d3934eb6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/d6a7014cacd5c1a1e50863eeb3044e68b7967c1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/d6a7014cacd5c1a1e50863eeb3044e68b7967c1a"}]},"branch":"refs/heads/master"},"b2518c58bcf500850ef475a4660d4afffb297d6f":{"kind":"REWORK","_number":5,"created":"2023-02-10 12:09:58.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/57/873357/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/57/873357/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/5"}}},"commit":{"parents":[{"commit":"2a7f392a80572ef69ce613fa80b1392b5558d869","subject":"Pass api_path thru etcd3gw.client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/2a7f392a80572ef69ce613fa80b1392b5558d869"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-10 09:38:59.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-10 12:09:48.000000000","tz":60},"subject":"Fix the docs job","message":"Fix the docs job\n\nDisable installing the current package in development mode\nwith develop mode.\n\nChange-Id: I05ab198c2cdec2683249caf9e591221d3934eb6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/b2518c58bcf500850ef475a4660d4afffb297d6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/b2518c58bcf500850ef475a4660d4afffb297d6f"}]},"branch":"refs/heads/master"},"1a2ddf8a91b666afd6f030736fb1b95ecc3f9a69":{"kind":"REWORK","_number":6,"created":"2023-02-10 13:08:38.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/57/873357/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/57/873357/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/57/873357/6"}}},"commit":{"parents":[{"commit":"2a7f392a80572ef69ce613fa80b1392b5558d869","subject":"Pass api_path thru etcd3gw.client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/2a7f392a80572ef69ce613fa80b1392b5558d869"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-10 09:38:59.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-10 13:07:30.000000000","tz":60},"subject":"Fix the docs job","message":"Fix the docs job\n\nChange-Id: I05ab198c2cdec2683249caf9e591221d3934eb6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/1a2ddf8a91b666afd6f030736fb1b95ecc3f9a69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/1a2ddf8a91b666afd6f030736fb1b95ecc3f9a69"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
