)]}'
{"id":"openstack%2Fetcd3gw~991968","triplet_id":"openstack%2Fetcd3gw~master~Ia1ba88593e1261e7f3dd0f37178dea661c3cbcd0","project":"openstack/etcd3gw","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-08 11:02:29.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-08 14:58:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia1ba88593e1261e7f3dd0f37178dea661c3cbcd0","subject":"Fix duplicate encode","status":"MERGED","created":"2026-06-06 15:08:20.000000000","updated":"2026-06-08 14:59:52.000000000","submitted":"2026-06-08 14:58:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991968","meta_rev_id":"5b80ea2981e44d637dd3ddf5a01e1540009da124","_number":991968,"virtual_id_number":991968,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-08 14:58:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-06-08 11:02:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-06-08 11:02:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-06 15:33:14.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-06-06 16:02:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b81bd65c67017eda827bdaf414eb47e5530c538","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-06 15:08:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b881b25b2cf2d34730f11d5f932726382bfb5b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 16:02:52.000000000","message":"Patch Set 1: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7bb21495a7c4f6ea2afedcef8f3a194\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75b0b1e1143e472b95c8a9e845635b61 : SUCCESS in 53m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/038a3da0bee447068026bdc0356de0e3 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fd2e0ab687241288a2918b6e56558a0 : FAILURE in 2m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/265b436979ed464f8c7e7adcf622e1d7 : SUCCESS in 2m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/867061e2a80248a1ab679d6735f47cbe : SUCCESS in 1m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9fe12d0527d24847ac242c31cb3ad7ca : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a4388883add48ea8a16cca5efb875c8 : SUCCESS in 2m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ba4ecf2723d41c9a40a420591c7c22e : SUCCESS in 2m 49s\n- etcd3gw-py312-functional https://zuul.opendev.org/t/openstack/build/f4a55e76863a49319af5e4db676631c6 : SUCCESS in 2m 50s\n\nWarning:\n  Comments left for invalid file etcd3gw/tests/test_client.py","accounts_in_message":[],"_revision_number":1},{"id":"687508a0f4ac9bde0b703fd6dc9e2b39662d8a4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-08 11:02:23.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"82c3ee7cec830992475d37c248bf1b117b26bb13","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-08 11:02:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ca2c9f364cf1d0f8f004e1553f08eb41fffc1ae1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 13:00:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9df33d3c878e4070b51e96ef24a20ba5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/12d133aca50a49e4bcaf4a876e254fa5 : SUCCESS in 1h 49m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a04bd9ffe4704ba9b9c6ccb01f149440 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bafbb3c603e4cff99e4f4e21ded3298 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d21487f7c7e44ac4b94719558b67d70b : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3d223f3053084238898a12a798cf5759 : SUCCESS in 3m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f0699681989242dc97c208f8908a20e2 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6a837f6522b408aa3b07a0b2bcc5979 : SUCCESS in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cac248058fa54fa8ac2d6e55f96eca70 : SUCCESS in 2m 59s\n- etcd3gw-py312-functional https://zuul.opendev.org/t/openstack/build/2210f60a3caa47bcb33b2b49674dc654 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":2},{"id":"2e5469978e2d803a4ffd8b07b46d618090d24d36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 13:01:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3a940b168254b751e11a495acd5742719e791c81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 14:58:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f2f224c3ea54ce0b5d784f6e121fe23\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/63bab1b8060e48ac91603f012ff924ad : SUCCESS in 1h 51m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd55dfa7f3024eb38309633680110a24 : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b4757edb99d64127b8ae7c7cde8249dc : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d3d3b99f9d242259fea5561db9543ae : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/188b8acd7f1d4ced8d0334e32c63d01c : SUCCESS in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ec4b86930ab440aa05dd6a821c83923 : SUCCESS in 4m 28s\n- etcd3gw-py312-functional https://zuul.opendev.org/t/openstack/build/75f116ce022d4173862b2ca18fc2a6fe : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":2},{"id":"b0f9e49c3c62e76bc9261e277c9edd37a309cc5a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 14:58:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5b80ea2981e44d637dd3ddf5a01e1540009da124","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 14:59:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9594818b234b4265904f9dcbf7d875d3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/739c351c5c104842812f7cc179947263 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ebdc351fd351e86d26315d28d99afcbe136ad376","revisions":{"13d46d81880375439fb2add66bd444533f9d4369":{"kind":"REWORK","_number":1,"created":"2026-06-06 15:08:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/991968/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/68/991968/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/68/991968/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/68/991968/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/68/991968/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/68/991968/1"}}},"commit":{"parents":[{"commit":"4d8dd1435515eac95c1b78b589a6bdbcf39d4d33","subject":"Merge \"typing: Use typed dicts return values\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/4d8dd1435515eac95c1b78b589a6bdbcf39d4d33"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-06 15:07:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-06 15:07:16.000000000","tz":540},"subject":"Fix duplicate encode","message":"Fix duplicate encode\n\nFix the regression caused by 90d59973442b953a3db6fb947042e68b93e3077f ,\ncurrently affecting range_end in watch_prefix function. This is now\ncausing occasional failure in taskflow due to wrongly limited watch\ntarget.\n\nChange-Id: Ia1ba88593e1261e7f3dd0f37178dea661c3cbcd0\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/13d46d81880375439fb2add66bd444533f9d4369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/13d46d81880375439fb2add66bd444533f9d4369"}]},"branch":"refs/heads/master"},"ebdc351fd351e86d26315d28d99afcbe136ad376":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2026-06-08 11:02:23.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/991968/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/68/991968/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/68/991968/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/68/991968/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/68/991968/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/68/991968/2"}}},"commit":{"parents":[{"commit":"e7e445a02155ea76084978504b6fb5a700c8b523","subject":"typing: Fix issues introduced by recent types-requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/e7e445a02155ea76084978504b6fb5a700c8b523"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-06 15:07:16.000000000","tz":540},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-08 11:01:39.000000000","tz":60},"subject":"Fix duplicate encode","message":"Fix duplicate encode\n\nFix the regression caused by 90d59973442b953a3db6fb947042e68b93e3077f,\ncurrently affecting range_end in watch_prefix function. This is now\ncausing occasional failure in taskflow due to wrongly limited watch\ntarget.\n\nChange-Id: Ia1ba88593e1261e7f3dd0f37178dea661c3cbcd0\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/ebdc351fd351e86d26315d28d99afcbe136ad376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/ebdc351fd351e86d26315d28d99afcbe136ad376"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
