)]}'
{"id":"openstack%2Fetcd3gw~993867","triplet_id":"openstack%2Fetcd3gw~master~I9c8c0ed1c1fcea3602655dada3f27e6516dff172","project":"openstack/etcd3gw","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-10-02 06:51:14.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I9c8c0ed1c1fcea3602655dada3f27e6516dff172","subject":"Add cross tests with oslo.cache and tooz","status":"MERGED","created":"2026-06-18 03:45:16.000000000","updated":"2026-10-02 09:32:22.000000000","submitted":"2026-10-02 09:30:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993867","meta_rev_id":"e7c7b4dbb2d34f551e0612dd4282515dd8bd5e29","_number":993867,"virtual_id_number":993867,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-10-02 09:30:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-10-02 02:54:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-10-02 02:54:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-18 04:53:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-18 09:40:06.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"}],"messages":[{"id":"5770bda29bb80a467359829a34e343af9e3e1252","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-18 03:45:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5da7f022cca9ea650a0bb5eeb3b86bc0004cd32d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-18 03:52:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3ce919a088256e20041443c5259bc3c3aa621689","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 04:53:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6df484d23aa45d781a83dc6d2dddf02\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8e1699998271408aabf3c0b7248081f6 : SUCCESS in 58m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5df8c86f4544444a930d602d6f639839 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59219cee5e6f4acc8401ec11c2689e9f : SUCCESS in 1m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee20bb41e12f48eca56d0d881b9637c6 : SUCCESS in 1m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d538a12f4d86494c9b6f07f566f5e0f0 : SUCCESS in 1m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e091c73b403c42c69d89409116117a16 : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39aff63d344c4720a3e9bc8cb9632b82 : SUCCESS in 2m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30112519cca949769d7798ece14da1c1 : SUCCESS in 2m 39s\n- etcd3gw-py312-functional https://zuul.opendev.org/t/openstack/build/1af32885b4b447eebbde6f6be59b43cd : SUCCESS in 3m 15s\n- etcd3gw-oslo.cache-functional https://zuul.opendev.org/t/openstack/build/6127a5c758504b46a2120b8440099c99 : RETRY_LIMIT in 2m 26s\n- etcd3gw-tooz-tox-py311 https://zuul.opendev.org/t/openstack/build/4db7975519254abd9d3f843f0024c190 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":2},{"id":"de09cdcf5258749a46cb7798166ff5a3e35bd0b9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-18 09:40:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aab8874400e65993566cd9bd673662849044e511","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 11:27:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a58111ec38442929cfbf8b121ffdf6e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9cce762efe943f8a89edeb192c4af9f : SUCCESS in 1h 41m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/110558b27bb0446883dbbdb646d50413 : SUCCESS in 2m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/170de24d0034495ea1baa42bb45c9e1a : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a2e47ce1384640a48e3db6ce933d993f : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4964b550ebae4ca7b4f5d05f8841de7b : SUCCESS in 3m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/67d7f64bfc2e4f9a901f9ec0573c12ab : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7752536d34d14db1ac7c48cc8373b3c5 : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a5a06c46abe432d8461d294b32d238b : SUCCESS in 2m 51s\n- etcd3gw-py312-functional https://zuul.opendev.org/t/openstack/build/dac6e5f199ca4c2a8572fc3a4902dbab : SUCCESS in 2m 06s\n- etcd3gw-oslo.cache-functional https://zuul.opendev.org/t/openstack/build/c533c5851c8245ad8360ada2625db4ef : RETRY_LIMIT in 3m 24s\n- etcd3gw-tooz-tox-py311 https://zuul.opendev.org/t/openstack/build/7f46c2d46af54f568ba73bcf5b762c46 : SUCCESS in 12m 06s","accounts_in_message":[],"_revision_number":2},{"id":"99dcc8651756ebb403d5eb05ed5e6b2d2760caf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-25 15:42:16.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"f3adf1841d341ef89572c9bc1e001c71b2bd658f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-25 16:51:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"18ee4284b3779eb5aa8c614be5c6f3ae1f6a34f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-25 16:54:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3df2cec7788c54274008f1e002176f85d2cbb361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 16:55:09.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/etcd3gw on branch master.  The\nproblem was:\n\n  Job etcd3gw-tooz-tox-py311 not defined\n\nThe problem appears in the \"opendev.org/openstack/etcd3gw\" project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - lib-forward-testing-python3\n        - openstack-python3-jobs\n        - openstack-python3-next-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n  ...\n\n  in \"openstack/etcd3gw/.zuul.yaml@master\", line 42","accounts_in_message":[],"_revision_number":5},{"id":"0f61a2966418eb3519b3ba299d7acf680acfc928","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-26 10:01:50.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"16d474b586ffc7b5b7dc13f659508e86e9496aab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-26 11:03:18.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/63f7c5dc723247e78193ab5d264fbeb6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc96e46a8f0a4cc987f14544f470b0b6 : SUCCESS in 1h 00m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0769fefaf1314ae4a054f8032b5fc0bb : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ceb8759a40b4a28ba5c497dd76b4586 : SUCCESS in 3m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d8958ba56ea8429db67892a378870ab6 : SUCCESS in 5m 15s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/8f2c468ae2984fc8be687ea6883429ec : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f12097e336349618d7f6785e3ffb32b : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f22b4841f43b4eae8bd8f1cdcea7bdf9 : SUCCESS in 1m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d200aa1deeba494988050ffe8f7e5c16 : SUCCESS in 2m 35s\n- etcd3gw-py314-functional https://zuul.opendev.org/t/openstack/build/dc58cf78d58344c6b4000b7354fbee8d : SUCCESS in 5m 28s\n- etcd3gw-oslo.cache-functional https://zuul.opendev.org/t/openstack/build/64e8b1df77af4c54a77174bf663ae897 : SUCCESS in 4m 25s\n- etcd3gw-tooz-tox-py314 https://zuul.opendev.org/t/openstack/build/0fa8119a97074df79eafb7d098317358 : FAILURE in 5m 30s","accounts_in_message":[],"_revision_number":6},{"id":"60a5ad2203b536c990536c74657c8e4c624ca09a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-26 16:32:15.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"a396e80e333af6fbdffa5befe1395d39da69eb31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-26 16:34:00.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"24ec23588bae36937410c1ce5116672ba8555f18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-26 18:13:18.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a7cef05381c49dd98c1b9c2471d448b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba694ae59732496696b8393508eaf44b : SUCCESS in 1h 33m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feea0e65bfc4471db0f8e739342641a8 : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/39eb2c0d16eb487495d1a272336b48a6 : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e28b677958eb403f993bcca25da6aa60 : SUCCESS in 6m 12s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/1a557bff40d146e88a984e845739ea08 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecdf805a3a97467f8393b324948c16c2 : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71f58c0cb10e459a88985bad9ecd9114 : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e7cc14287ce4b8ebd8460ead0678af3 : SUCCESS in 6m 03s\n- etcd3gw-py314-functional https://zuul.opendev.org/t/openstack/build/b74c0099ee5246bcaaac4c4aae615e3f : SUCCESS in 5m 32s\n- etcd3gw-oslo.cache-functional https://zuul.opendev.org/t/openstack/build/4cd21a35aee14f0881d938d3600f597d : RETRY_LIMIT in 4m 27s\n- etcd3gw-tooz-tox-py3 https://zuul.opendev.org/t/openstack/build/5f76fde6fae6475991ba37e342709747 : RETRY_LIMIT in 5m 09s","accounts_in_message":[],"_revision_number":8},{"id":"f45d5234b2d61cbf98e4299c5ef5ebacfff5c49a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-26 19:47:34.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"82d5530dc22576203c04fda571e42e6866b8fd27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-26 21:26:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2475c88265d4e9787b70e46c37fe8bd\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/301d82104fa84947bc312a64ae91f36e : SUCCESS in 1h 31m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6242ac1be21a4696b50398f4870cb0d7 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c6416b8640044498ea51a1099a10381 : SUCCESS in 3m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6259d056c0934af7a3282c533001961f : SUCCESS in 4m 02s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/c1d7e59a38954564991bc653053061c1 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f9c2ca40a89450d98cbaec8fa190d07 : SUCCESS in 3m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f390e8602fa84ccea1eea1df0a82883f : SUCCESS in 2m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67a2403d06d34c44b71e3dc6a0a970fe : SUCCESS in 3m 36s\n- etcd3gw-py314-functional https://zuul.opendev.org/t/openstack/build/160a42fdc4a247c3b32ba7eeba0cb99d : SUCCESS in 4m 27s\n- etcd3gw-oslo.cache-functional https://zuul.opendev.org/t/openstack/build/8a8b507d2a964336b1add41e4204886a : SUCCESS in 2m 46s\n- etcd3gw-tooz-tox-py3 https://zuul.opendev.org/t/openstack/build/dd2d127b2bfc4909a0d37bddb4b692dc : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":9},{"id":"467850453d8f46243661b3e2bea1dd1efe17f0ff","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-10-02 02:54:36.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"9f81bb18141669c7fbd0c65444697ba4ca68f969","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 02:54:47.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/efbc3aa9255c4ef39a9cf15c22adefc9\n\n\nWarning:\n  Change 1007587,1 in project openstack/tooz does not share a change queue with 993867 in project openstack/etcd3gw","accounts_in_message":[],"_revision_number":9},{"id":"12b1a634843eacd93a7d4ca11fd60546ba9d1432","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-10-02 06:51:14.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"61b6f8f34fd220a962345616228f9b62296bc038","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 08:43:38.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31eac56be5354b2485193c567b714a1e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/748e9cd1b4604a42a72c3bc0cbf74ee8 : SUCCESS in 1h 49m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a52b6ece9674676b4536c09a4aa1290 : SUCCESS in 2m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b6e734731c894d4db0c3e7faef50200a : SUCCESS in 3m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bb97ed76a4f144138e3c6483ce493e72 : SUCCESS in 6m 23s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/4ac90ced42f24ac7ac2f27788517dae3 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9ef565737b84f5aa7b924c688654081 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81b3aa17d5374f529c0d395122340d81 : SUCCESS in 2m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b919ad5567e547a2a6d603441a502f7b : SUCCESS in 5m 25s\n- etcd3gw-py314-functional https://zuul.opendev.org/t/openstack/build/8b87d09e3f964bfd80fef6a4bea16516 : SUCCESS in 5m 53s\n- etcd3gw-oslo.cache-functional https://zuul.opendev.org/t/openstack/build/eebb10e465dd4b938aee7443ed3828f0 : SUCCESS in 3m 25s\n- etcd3gw-tooz-tox-py3 https://zuul.opendev.org/t/openstack/build/ff0589ef886b4fc584383b43bbd55e59 : SUCCESS in 7m 41s","accounts_in_message":[],"_revision_number":9},{"id":"5aa8eeb56135cc989b8a61974a68cc8ff635f25d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 08:44:12.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"69a37282d6775384999f4ed181e9df8d67279e1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 09:30:25.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4114aad2f14448c998ba47ecf7662eb0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/710801d5271a4650b70af54f09157c11 : SUCCESS in 45m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a9d5e38eacd412daf88048063d971b6 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f877abd93ce47209aeeb045942792cf : SUCCESS in 2m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3706a01297c546afa93fbe4dec492258 : SUCCESS in 3m 46s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/26546c3a20164096a0fc397d2c3fa696 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17b963223af54f9ca1fb3b4998cf029b : SUCCESS in 2m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f83b1e967a704544b26d9d961ee42b4c : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b26f82bc4f5482988f71132032f22aa : SUCCESS in 3m 29s\n- etcd3gw-py314-functional https://zuul.opendev.org/t/openstack/build/d2b7858ea4de4cad82e920ea0142d438 : SUCCESS in 3m 51s\n- etcd3gw-oslo.cache-functional https://zuul.opendev.org/t/openstack/build/0c49c0395b8446c29f6b4fadbeb03169 : SUCCESS in 4m 02s\n- etcd3gw-tooz-tox-py3 https://zuul.opendev.org/t/openstack/build/9702be0ccbd44c3da58713f66124cce6 : SUCCESS in 7m 18s","accounts_in_message":[],"_revision_number":9},{"id":"a6c0d14bd11d625a4f37184b1a18a93899195b37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 09:30:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"e7c7b4dbb2d34f551e0612dd4282515dd8bd5e29","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 09:32:22.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f050651f1d5f4396b19dd09eea77fac3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/436633dffe1246d98e704caeeff3a728 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87b2256ed04a4f439071b0b7164ce045 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"fce27f745334a26011ec9b6e444e678d94edddda","revisions":{"9c50859d43f845c2c44333abdc1b0f3d09fc8770":{"kind":"REWORK","_number":1,"created":"2026-06-18 03:45:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/993867/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/67/993867/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/1"}}},"commit":{"parents":[{"commit":"ebdc351fd351e86d26315d28d99afcbe136ad376","subject":"Fix duplicate encode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/ebdc351fd351e86d26315d28d99afcbe136ad376"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-18 03:43:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-18 03:43:09.000000000","tz":540},"subject":"Add cross tests with oslo.cache and tooz","message":"Add cross tests with oslo.cache and tooz\n\noslo.cache and tooz are major consumers of this library. Add cross\ntests so that we catch potential problems early.\n\nChange-Id: I9c8c0ed1c1fcea3602655dada3f27e6516dff172\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/9c50859d43f845c2c44333abdc1b0f3d09fc8770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/9c50859d43f845c2c44333abdc1b0f3d09fc8770"}]},"branch":"refs/heads/master"},"c90bb0ba2aacf8264fd918ddc7a528849828a9ee":{"kind":"REWORK","_number":2,"created":"2026-06-18 03:52:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/993867/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/67/993867/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/2"}}},"commit":{"parents":[{"commit":"ebdc351fd351e86d26315d28d99afcbe136ad376","subject":"Fix duplicate encode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/ebdc351fd351e86d26315d28d99afcbe136ad376"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-18 03:43:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-18 03:52:22.000000000","tz":540},"subject":"Add cross tests with oslo.cache and tooz","message":"Add cross tests with oslo.cache and tooz\n\noslo.cache and tooz are major consumers of this library. Add cross\ntests so that we catch potential problems early.\n\nChange-Id: I9c8c0ed1c1fcea3602655dada3f27e6516dff172\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/c90bb0ba2aacf8264fd918ddc7a528849828a9ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/c90bb0ba2aacf8264fd918ddc7a528849828a9ee"}]},"branch":"refs/heads/master"},"27b6e45cda57688c5cfab8d368e7c7d58a654ff6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-25 15:42:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/993867/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/67/993867/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/3"}}},"commit":{"parents":[{"commit":"6f3ab0974a27adbf1946080eeb6ddd0b9d420f0c","subject":"Declare Python 3.15 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/6f3ab0974a27adbf1946080eeb6ddd0b9d420f0c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-18 03:43:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 15:42:16.000000000","tz":0},"subject":"Add cross tests with oslo.cache and tooz","message":"Add cross tests with oslo.cache and tooz\n\noslo.cache and tooz are major consumers of this library. Add cross\ntests so that we catch potential problems early.\n\nChange-Id: I9c8c0ed1c1fcea3602655dada3f27e6516dff172\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/27b6e45cda57688c5cfab8d368e7c7d58a654ff6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/27b6e45cda57688c5cfab8d368e7c7d58a654ff6"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"ebdc351fd351e86d26315d28d99afcbe136ad376","ours":"c90bb0ba2aacf8264fd918ddc7a528849828a9ee","theirs":"6f3ab0974a27adbf1946080eeb6ddd0b9d420f0c","merge_strategy":"recursive","contains_conflicts":false}},"2db026a2c47698d35ec0991a6e8ed3c3fc4c885a":{"kind":"REWORK","_number":4,"created":"2026-09-25 16:51:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/993867/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/67/993867/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/4"}}},"commit":{"parents":[{"commit":"6f3ab0974a27adbf1946080eeb6ddd0b9d420f0c","subject":"Declare Python 3.15 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/6f3ab0974a27adbf1946080eeb6ddd0b9d420f0c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-18 03:43:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 16:51:44.000000000","tz":540},"subject":"Add cross tests with oslo.cache and tooz","message":"Add cross tests with oslo.cache and tooz\n\noslo.cache and tooz are major consumers of this library. Add cross\ntests so that we catch potential problems early.\n\nChange-Id: I9c8c0ed1c1fcea3602655dada3f27e6516dff172\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/2db026a2c47698d35ec0991a6e8ed3c3fc4c885a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/2db026a2c47698d35ec0991a6e8ed3c3fc4c885a"}]},"branch":"refs/heads/master"},"71d56852c29a70ec0eaaccbe8ee42a01b090e17c":{"kind":"REWORK","_number":5,"created":"2026-09-25 16:54:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/993867/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/67/993867/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/5"}}},"commit":{"parents":[{"commit":"6f3ab0974a27adbf1946080eeb6ddd0b9d420f0c","subject":"Declare Python 3.15 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/6f3ab0974a27adbf1946080eeb6ddd0b9d420f0c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-18 03:43:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 16:54:28.000000000","tz":540},"subject":"Add cross tests with oslo.cache and tooz","message":"Add cross tests with oslo.cache and tooz\n\noslo.cache and tooz are major consumers of this library. Add cross\ntests so that we catch potential problems early.\n\nChange-Id: I9c8c0ed1c1fcea3602655dada3f27e6516dff172\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/71d56852c29a70ec0eaaccbe8ee42a01b090e17c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/71d56852c29a70ec0eaaccbe8ee42a01b090e17c"}]},"branch":"refs/heads/master"},"06b9719fa0a1687148c36331294196ea4e67d30d":{"kind":"REWORK","_number":6,"created":"2026-09-26 10:01:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/993867/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/67/993867/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/6"}}},"commit":{"parents":[{"commit":"a6e1eb90abf62d85f3ad0cdced3b4efc6233636c","subject":"Move functional tests to py314","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/a6e1eb90abf62d85f3ad0cdced3b4efc6233636c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-18 03:43:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-26 10:01:38.000000000","tz":540},"subject":"Add cross tests with oslo.cache and tooz","message":"Add cross tests with oslo.cache and tooz\n\noslo.cache and tooz are major consumers of this library. Add cross\ntests so that we catch potential problems early.\n\nChange-Id: I9c8c0ed1c1fcea3602655dada3f27e6516dff172\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/06b9719fa0a1687148c36331294196ea4e67d30d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/06b9719fa0a1687148c36331294196ea4e67d30d"}]},"branch":"refs/heads/master"},"0ef3fffb3d282f3bed0d2d002ffdede3b386ce77":{"kind":"REWORK","_number":7,"created":"2026-09-26 16:32:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/993867/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/67/993867/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/7"}}},"commit":{"parents":[{"commit":"a6e1eb90abf62d85f3ad0cdced3b4efc6233636c","subject":"Move functional tests to py314","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/a6e1eb90abf62d85f3ad0cdced3b4efc6233636c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-18 03:43:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-26 16:31:54.000000000","tz":540},"subject":"Add cross tests with oslo.cache and tooz","message":"Add cross tests with oslo.cache and tooz\n\noslo.cache and tooz are major consumers of this library. Add cross\ntests so that we catch potential problems early.\n\nDepends-on: https://review.opendev.org/1007587\nChange-Id: I9c8c0ed1c1fcea3602655dada3f27e6516dff172\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/0ef3fffb3d282f3bed0d2d002ffdede3b386ce77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/0ef3fffb3d282f3bed0d2d002ffdede3b386ce77"}]},"branch":"refs/heads/master"},"6aefbd8544b92ddb70ac832a053e2562a5a25596":{"kind":"REWORK","_number":8,"created":"2026-09-26 16:34:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/993867/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/67/993867/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/8"}}},"commit":{"parents":[{"commit":"a6e1eb90abf62d85f3ad0cdced3b4efc6233636c","subject":"Move functional tests to py314","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/a6e1eb90abf62d85f3ad0cdced3b4efc6233636c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-18 03:43:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-26 16:33:49.000000000","tz":540},"subject":"Add cross tests with oslo.cache and tooz","message":"Add cross tests with oslo.cache and tooz\n\noslo.cache and tooz are major consumers of this library. Add cross\ntests so that we catch potential problems early.\n\nDepends-on: https://review.opendev.org/1007587\nChange-Id: I9c8c0ed1c1fcea3602655dada3f27e6516dff172\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/6aefbd8544b92ddb70ac832a053e2562a5a25596"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/6aefbd8544b92ddb70ac832a053e2562a5a25596"}]},"branch":"refs/heads/master"},"fce27f745334a26011ec9b6e444e678d94edddda":{"kind":"REWORK","_number":9,"created":"2026-09-26 19:47:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/993867/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/67/993867/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/67/993867/9"}}},"commit":{"parents":[{"commit":"a6e1eb90abf62d85f3ad0cdced3b4efc6233636c","subject":"Move functional tests to py314","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/a6e1eb90abf62d85f3ad0cdced3b4efc6233636c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-18 03:43:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-26 19:47:21.000000000","tz":540},"subject":"Add cross tests with oslo.cache and tooz","message":"Add cross tests with oslo.cache and tooz\n\noslo.cache and tooz are major consumers of this library. Add cross\ntests so that we catch potential problems early.\n\nDepends-on: https://review.opendev.org/1007587\nChange-Id: I9c8c0ed1c1fcea3602655dada3f27e6516dff172\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/fce27f745334a26011ec9b6e444e678d94edddda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/fce27f745334a26011ec9b6e444e678d94edddda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
