)]}'
{"id":"openstack%2Fetcd3gw~994263","triplet_id":"openstack%2Fetcd3gw~master~Ic1bdd98ad0e3f35bf124d2740693ec2f50ad722f","project":"openstack/etcd3gw","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-24 14:25:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic1bdd98ad0e3f35bf124d2740693ec2f50ad722f","subject":"Do not install code to build release notes","status":"MERGED","created":"2026-06-22 12:19:07.000000000","updated":"2026-06-24 14:26:13.000000000","submitted":"2026-06-24 14:25:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994263","meta_rev_id":"281f70cfb070e5a53a8820067e374e5f0552dc47","_number":994263,"virtual_id_number":994263,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-24 14:25:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-06-24 12:31:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-06-24 12:31:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-22 14:11:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-24 12:31:32.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"782f2d73e079e48c6361f37421d205de96a75e64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-22 12:19:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d415ea32d24e1578300ccf84ff918b284d5b33f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 14:11:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b1ec4f16b4240c0afdb64c3d91d03a7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac648aa917004dc4bd985587f9fba723 : SUCCESS in 1h 46m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6df8d92678c94fa8872a693431d21e79 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f83f20ff19ce41558468623f5f093b35 : FAILURE in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/09f3e2237a394333a217c473cd546170 : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/28ee6d958b6649b29d1155e7d892eb85 : SUCCESS in 3m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ca495c59bcf84321b223c11a515d942d : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0feadf719924e5488727dfcd20f4226 : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1af7bc6cb7ca424e8f9cdc6ceec98320 : SUCCESS in 4m 03s\n- etcd3gw-py312-functional https://zuul.opendev.org/t/openstack/build/70d3224708a0417ea5aba02a0a9d969a : SUCCESS in 2m 38s\n\nWarning:\n  Comments left for invalid file etcd3gw/__init__.py\n  Comments left for invalid file etcd3gw/utils.py","accounts_in_message":[],"_revision_number":1},{"id":"96f5d1c32829700461dd489e81acf96b463ae481","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-23 17:54:00.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"3417eff52f10103bcfe00d1c79e0c24399d3f5cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 18:47:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bfca1c3e3e24bc695b6cc14d5b14de7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b64038fd0cba422fbc9259af234eca19 : SUCCESS in 52m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8bf62e2ff6a42ffbdd68165b67a1d9a : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa7a9926139b49e3bafc609fbaf44b36 : SUCCESS in 3m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac627f36bf9643fba266e99feeead8f5 : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4925826a13e941c78c4c5922a0e03831 : SUCCESS in 2m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c4c07bfbb57e4131aabc64579625af72 : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50582394fa074be19c2899f05fae8985 : SUCCESS in 1m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5f98b616e7e417d90cf7ca43d12a5cf : SUCCESS in 2m 52s\n- etcd3gw-py312-functional https://zuul.opendev.org/t/openstack/build/52b9ab1c6e344dc5b193bcef1dfb978c : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":2},{"id":"3136709ea28a6a8b53455181cfe4dc70d64af4dd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-24 12:31:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a0d3b243d43d1542cf04b8673a272e831548efc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 12:31:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"76c2ea4c927c4cfa1be2f8b564afb3ecf1f9cd6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 14:25:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0f714ec87a14b8bb9b85217e7b89d07\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8c1a85f74232492db1f94ab745657acd : SUCCESS in 1h 48m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ece44735d1c4787826d20ea7d42c006 : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4a13a1dd07644e7abe7a8db6b9599c5 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da3bd191fe1c4721bb108fde082fb86e : SUCCESS in 1m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e6039bb5b3e4479ba899d4fc9a543a8e : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0ea16da7aff4910aeb5cb42a156d7dd : SUCCESS in 4m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d41c79ab12b47fca44c05884a291b7a : SUCCESS in 2m 36s\n- etcd3gw-py312-functional https://zuul.opendev.org/t/openstack/build/5037716b38e9455d837fb713306a9243 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":2},{"id":"e4fe5086bafeac32f11acf21e21a65c854e55a35","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 14:25:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"281f70cfb070e5a53a8820067e374e5f0552dc47","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 14:26:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9002948408b04be4bcca893deb6a8780\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/862af51f425d4ec6bcb0f6d2064b8853 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84e7a2ee89e64e85a45ab080fa63d7c7 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0f7a6ca20d67e44874a04f2dcaae7afdac79ab26","revisions":{"d10c10dbfd69f9ace08cd650b9dde9c4a5edb9b7":{"kind":"REWORK","_number":1,"created":"2026-06-22 12:19:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/994263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/63/994263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/63/994263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/63/994263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/63/994263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/63/994263/1"}}},"commit":{"parents":[{"commit":"ebdc351fd351e86d26315d28d99afcbe136ad376","subject":"Fix duplicate encode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/ebdc351fd351e86d26315d28d99afcbe136ad376"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-22 12:18:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-22 12:19:00.000000000","tz":540},"subject":"Do not install code to build release notes","message":"Do not install code to build release notes\n\n... because we require only doc requirements.\n\nAlso remove explicit usage of requirements.txt because it is now\nautomatically detected and installed.\n\nChange-Id: Ic1bdd98ad0e3f35bf124d2740693ec2f50ad722f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/d10c10dbfd69f9ace08cd650b9dde9c4a5edb9b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/d10c10dbfd69f9ace08cd650b9dde9c4a5edb9b7"}]},"branch":"refs/heads/master"},"0f7a6ca20d67e44874a04f2dcaae7afdac79ab26":{"kind":"REWORK","_number":2,"created":"2026-06-23 17:54:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/994263/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/etcd3gw","ref":"refs/changes/63/994263/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/63/994263/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/63/994263/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/etcd3gw refs/changes/63/994263/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/etcd3gw refs/changes/63/994263/2"}}},"commit":{"parents":[{"commit":"ebdc351fd351e86d26315d28d99afcbe136ad376","subject":"Fix duplicate encode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/ebdc351fd351e86d26315d28d99afcbe136ad376"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-22 12:18:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-23 17:53:50.000000000","tz":540},"subject":"Do not install code to build release notes","message":"Do not install code to build release notes\n\n... because we require only doc requirements.\n\nAlso remove explicit usage of requirements.txt because it is now\nautomatically detected and installed.\n\nChange-Id: Ic1bdd98ad0e3f35bf124d2740693ec2f50ad722f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/0f7a6ca20d67e44874a04f2dcaae7afdac79ab26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/etcd3gw/commit/0f7a6ca20d67e44874a04f2dcaae7afdac79ab26"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
