)]}'
{"id":"openstack%2Ffreezer-api~1006691","triplet_id":"openstack%2Ffreezer-api~master~I83bf3eef74dca5be733bb68784c3edcf87884df9","project":"openstack/freezer-api","branch":"master","hashtags":[],"change_id":"I83bf3eef74dca5be733bb68784c3edcf87884df9","subject":"Add negative tests for user update and cross-project access","status":"MERGED","created":"2026-09-22 09:08:40.000000000","updated":"2026-09-22 10:56:06.000000000","submitted":"2026-09-22 10:55:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006691","meta_rev_id":"4cc56a6bd08f9f611b841d067391ac31bf0458c5","_number":1006691,"virtual_id_number":1006691,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-22 10:55:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-09-22 10:15:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-09-22 10:15:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-22 09:23:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-22 10:07:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3bd8edacc17df67f60077085ffaa1c9d70b1a1c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 09:08:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da37706ace43fcc6e37e9de7494137db6b869b50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 09:23:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07a16862c9874177957d1ec0fdf4b8e2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a9bb6b6778dd4389846b7086ae905529 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/17116c69c61e433b8797607c238d2c97 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/654bc0488950404d9d06d82d4c4f1df1 : SUCCESS in 11m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7406af30b99282a999b24370fba4667ae1f8a212","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 09:26:53.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9ba8f2445a5773d9db27464ca580a998c42b7843","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 09:39:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c3bc36ad5a24a36b476502975ea474e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bfee95d3eeae4be5b807cf2715ad1260 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/539f069ba59241b894ab29da88edf21b : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/857217dc386540478c44ca2c14a188c3 : SUCCESS in 10m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c49e8602c8361d921864ffd2c834248ee91288f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 10:07:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22029ccdf436446cb28718cccf72064c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdcda7759b894176b4687e3bec5984f8 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c7bbc11e73804eb49444a5a389fb09ea : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/586e7516338e49c8884b3a56ea65f3d2 : SUCCESS in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b992830dbe2d45a583ad4cfbc509635a : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e753fb3b21a48dd83cc1ca7623a87be : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7044be46374c4327af88b9e79d6f300e : SUCCESS in 4m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ca3f8981146340dca82b670d98ba6c4d : SUCCESS in 3m 34s\n- freezer-tempest-basic https://zuul.opendev.org/t/openstack/build/90d4b5d82672421ebe1ae3bfe0618a39 : SUCCESS in 19m 12s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fab0ba7916b04676a416d3013e878069 : SUCCESS in 33m 21s","accounts_in_message":[],"_revision_number":2},{"id":"d2d9b5cde58ee2f1da15af39782e3b76048df423","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 10:15:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"de46dbce17a2f304d17f4e13ddad30141644a35d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 10:16:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b0d86c4d646d090c1b2c3021700b57ea8732088e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 10:55:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac67004f6454494bb9c8e545b9dc845e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c43bbffab294f1faae8fca364e40ba6 : SUCCESS in 2m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4668721b95144c589a51071ccf835cf6 : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/259a0a5a29ad4e0fac697b77a527eff6 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2231e67fe5584cc3bc40a2a0c3e982fe : SUCCESS in 2m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4913bd886d084379ba234ec6de7399a3 : SUCCESS in 3m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/62842d11ffe44f398bb3825f8d4e43b7 : SUCCESS in 2m 34s\n- freezer-tempest-basic https://zuul.opendev.org/t/openstack/build/5815ba94b8ba4e23a34bd7abf1cdd2d0 : SUCCESS in 33m 42s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af0278d8ddaf42139af1767d95b5cf98 : SUCCESS in 26m 00s","accounts_in_message":[],"_revision_number":2},{"id":"9a7714d37d0684d0d713e55537825b4134b76062","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 10:55:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4cc56a6bd08f9f611b841d067391ac31bf0458c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 10:56:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1eac1151eff444ba9456b3ef0f4986e7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3d7e1e595674240a41f15b0cb33f966 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8118d6f6f1b36326e4310328de93b82fa2e6d8b5","revisions":{"f354d6e2840426eba2bb598718f94883f23ec6b3":{"kind":"REWORK","_number":1,"created":"2026-09-22 09:08:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/91/1006691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-api","ref":"refs/changes/91/1006691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-api refs/changes/91/1006691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-api refs/changes/91/1006691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-api refs/changes/91/1006691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-api refs/changes/91/1006691/1"}}},"commit":{"parents":[{"commit":"b0b73c715e7143b338fa3d9078b5f94226210527","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-api/commit/b0b73c715e7143b338fa3d9078b5f94226210527"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 09:08:22.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 09:08:22.000000000","tz":120},"subject":"Add negative tests for user update and cross-project access","message":"Add negative tests for user update and cross-project access\n\nWe were lacking proper negative testing of the API calls for\ncross-project access isolation, and that user_id update is always coming\nfrom the context and not the supplied doc content.\n\nAssisted-By: Gemini 3.8\nRelated-Bug: #2167261\nChange-Id: I83bf3eef74dca5be733bb68784c3edcf87884df9\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-api/commit/f354d6e2840426eba2bb598718f94883f23ec6b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-api/commit/f354d6e2840426eba2bb598718f94883f23ec6b3"}]},"branch":"refs/heads/master"},"8118d6f6f1b36326e4310328de93b82fa2e6d8b5":{"kind":"REWORK","_number":2,"created":"2026-09-22 09:26:53.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/91/1006691/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-api","ref":"refs/changes/91/1006691/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-api refs/changes/91/1006691/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-api refs/changes/91/1006691/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-api refs/changes/91/1006691/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-api refs/changes/91/1006691/2"}}},"commit":{"parents":[{"commit":"b0b73c715e7143b338fa3d9078b5f94226210527","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-api/commit/b0b73c715e7143b338fa3d9078b5f94226210527"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 09:08:22.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 09:26:53.000000000","tz":0},"subject":"Add negative tests for user update and cross-project access","message":"Add negative tests for user update and cross-project access\n\nWe were lacking proper negative testing of the API calls for\ncross-project access isolation, and that user_id update is always coming\nfrom the context and not the supplied doc content.\n\nAssisted-By: Gemini 3.8\nRelated-Bug: #2167261\nChange-Id: I83bf3eef74dca5be733bb68784c3edcf87884df9\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-api/commit/8118d6f6f1b36326e4310328de93b82fa2e6d8b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-api/commit/8118d6f6f1b36326e4310328de93b82fa2e6d8b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
