)]}'
{"id":"openstack%2Ffreezer-api~1006701","triplet_id":"openstack%2Ffreezer-api~master~I4c99f8d93e18ba88def46aeb9b535a2734cd2901","project":"openstack/freezer-api","branch":"master","hashtags":[],"change_id":"I4c99f8d93e18ba88def46aeb9b535a2734cd2901","subject":"Bind resources by project_id for Elasticsearch","status":"MERGED","created":"2026-09-22 10:02:50.000000000","updated":"2026-09-25 08:11:20.000000000","submitted":"2026-09-25 08:09:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006701","meta_rev_id":"44539606ab70b7c0ad012b618c06a908b56cc1de","_number":1006701,"virtual_id_number":1006701,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-25 08:09:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-09-25 07:28:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-09-25 07:28:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-22 10:15:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-22 10:37:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e80e0151519ed5f158849c96564bd33ef02d5d18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 10:02:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23986f20b2452a4e8dbffec3ba977517333a821c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 10:15:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c334f8753af6489a9d7db97690337aa2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3d47fe423901435c9861fbbb4bf1f081 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9fe0ef1aabed46a5aaf69f454f00ce24 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d01780b932d34c34a8032550f5238a1e : SUCCESS in 9m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"081d49a57e36bfd4b4bd58ba2cdcdd0443033af1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 10:37:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cdbaf961f9a4278904c34b66eb48d25\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41d0805e5f4843c1b35e1b52dcf899ef : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c6198dd913e946bd87aed5c4417f218c : SUCCESS in 5m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8699f4048d7e4d5084488131ba9adc17 : SUCCESS in 3m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d7d0f6b778b64f44a70022f85e460ac3 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75bd7d7d26784880ade5b8986f35e853 : SUCCESS in 2m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80190c4bfc054d129fbf1396cb7386ee : SUCCESS in 5m 34s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1d7e017cbcf6410fa9cad9c841822348 : SUCCESS in 4m 23s\n- freezer-tempest-basic https://zuul.opendev.org/t/openstack/build/1d90ba2bf41c4d23b575cf0d5eab6408 : SUCCESS in 32m 38s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6dd8405e02844b07919e9633e9f88170 : SUCCESS in 23m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1f1f4833cdf2e8710582c81dd3a2ce4a043010eb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-25 07:28:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f3d1aa95927fcff3468104318330af87f03ef80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 07:29:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"79aa332c0f20866e8fa8e5ca6887223d7be13323","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 08:09:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24a06ee207724126ae83d48b721d9ee5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d3a4de56ccf443abe3407fafcc93054 : SUCCESS in 2m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/77d92fdea6ea4f17baf2d2d8c6c8e14d : SUCCESS in 2m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/323bb6f28c8b45be8ebd6275f7e320c8 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/875c403cee5645acaf7de36aab14eebc : SUCCESS in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6fe4e6c4db54ec5a5e8adc76ded3444 : SUCCESS in 3m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/65905068a0114a3d9a93f748102ae397 : SUCCESS in 4m 02s\n- freezer-tempest-basic https://zuul.opendev.org/t/openstack/build/267cf8300e974e4ab01bf1cc6b4655a9 : SUCCESS in 18m 10s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91cff8a3591b44608aa21273df9e360e : SUCCESS in 35m 12s","accounts_in_message":[],"_revision_number":1},{"id":"b32e1c9ad64414e0bc66b271e2ca1269baf1ff74","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 08:09:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"44539606ab70b7c0ad012b618c06a908b56cc1de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 08:11:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d04e27ea389340678212a67feed35403\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19a02945d4b342bbb9a9810f28012cbf : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40ece7b122a06956c013ba7d9d4206c3039fee9d","revisions":{"40ece7b122a06956c013ba7d9d4206c3039fee9d":{"kind":"REWORK","_number":1,"created":"2026-09-22 10:02:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/01/1006701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-api","ref":"refs/changes/01/1006701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-api refs/changes/01/1006701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-api refs/changes/01/1006701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-api refs/changes/01/1006701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-api refs/changes/01/1006701/1"}}},"commit":{"parents":[{"commit":"8118d6f6f1b36326e4310328de93b82fa2e6d8b5","subject":"Add negative tests for user update and cross-project access","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-api/commit/8118d6f6f1b36326e4310328de93b82fa2e6d8b5"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 10:02:21.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 10:02:28.000000000","tz":120},"subject":"Bind resources by project_id for Elasticsearch","message":"Bind resources by project_id for Elasticsearch\n\nThis is a follow-up to the change [1] which has implemented mapping of\nall resources by project_id instead of the user_id.\n\nPrior change has ommited elasticsearch engine, as it\u0027s being deprecated.\nWe align the storage engines, as elasticsearch was not yet removed, and\nthis difference in behavior between engines is unacceptable.\n\nBinding resources by user_id is a consequence of v1 API, where binding\nby user_id was the way to enforce ownership back in the days.\n\n[1] https://review.opendev.org/c/openstack/freezer-api/+/988643\n\nRelated-Bug: #2167261\nChange-Id: I4c99f8d93e18ba88def46aeb9b535a2734cd2901\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-api/commit/40ece7b122a06956c013ba7d9d4206c3039fee9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-api/commit/40ece7b122a06956c013ba7d9d4206c3039fee9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
