)]}'
{"id":"openstack%2Ffreezer-tempest-plugin~1004420","triplet_id":"openstack%2Ffreezer-tempest-plugin~master~Ieeb625f895b3f73f4a0b7d38c231f7391dcd4c9e","project":"openstack/freezer-tempest-plugin","branch":"master","topic":"freezer/backups_update","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-08 16:42:09.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Ieeb625f895b3f73f4a0b7d38c231f7391dcd4c9e","subject":"Expect 404 return code for non-existent resource deletion","status":"MERGED","created":"2026-09-07 13:28:34.000000000","updated":"2026-09-08 17:29:46.000000000","submitted":"2026-09-08 17:29:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004420-freezer/backups_update","meta_rev_id":"40c0a1dd064c8ff4f722222a7dcd06badaeae1f6","_number":1004420,"virtual_id_number":1004420,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-08 17:29:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-09-08 16:42:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-09-08 16:42:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 14:27:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3b65d9e9c888215472ad9d434718d3f7b065112a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-07 13:28:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87b11f10c296194ae393d6057acdf240644c7a52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 14:27:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7dce168ac11541fbba0230681ad8d241\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fc991693a7248bdad17cad9c6cd0d37 : SUCCESS in 2m 07s\n- freezer-tempest-basic https://zuul.opendev.org/t/openstack/build/df09dd9a77f54c61a00c49e152b03ded : FAILURE in 37m 30s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd7f920ed7b64db8b6dddefb3198b51a : FAILURE in 20m 23s\n- freezer-tempest-client https://zuul.opendev.org/t/openstack/build/9ee6e5a3254444099e7f81508233c1db : SUCCESS in 35m 54s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/8a6a184302ee4fd084420513760cb472 : SUCCESS in 31m 42s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/318fe073cabc4255b70c946ec7e27b33 : SUCCESS in 19m 18s","accounts_in_message":[],"_revision_number":1},{"id":"bb6e293b406c2d70c97b6c9f732d4e0ed2571646","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-08 13:54:14.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"b9f36c4971325249311a6a8c0dcadc195e4a6f72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 14:37:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8db18e1e4e974d06899ff8c8b972745b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20b266502d4f4e4c87c5732bbc65e695 : SUCCESS in 2m 33s\n- freezer-tempest-client https://zuul.opendev.org/t/openstack/build/6210f4f5c792463a9d5aea29cb11d359 : SUCCESS in 36m 31s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/ded728d123a54420a8499cf40bba9e89 : SUCCESS in 41m 25s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/b65b2e78a7da4995849788849a2f6cad : SUCCESS in 36m 59s","accounts_in_message":[],"_revision_number":2},{"id":"bbfc40da2e671de2d6d163239ca95292341ce883","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-08 16:42:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"abb9e5f9af393f3d329398e2dfe0854745dba564","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 16:42:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6bf158b456c1834bd21138e328df4129ef846cbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 17:29:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1adf6773b1014baf984d154c5d32dfc4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97a2248590fd418e98b8362a9b3ce429 : SUCCESS in 2m 28s\n- freezer-tempest-client https://zuul.opendev.org/t/openstack/build/68f0230fd88b496c9f862810abc714a7 : SUCCESS in 22m 52s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/330bba81c3ef489f831c9c9ea2abc8df : SUCCESS in 45m 28s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/f0306ece39c24ea8b3fadd640f17b0c0 : SUCCESS in 34m 24s","accounts_in_message":[],"_revision_number":2},{"id":"40c0a1dd064c8ff4f722222a7dcd06badaeae1f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 17:29:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e5d62a3a7abebd1569e06d27a50879d78f3ea218","revisions":{"b2a919410233e389573da1c41ead0c0088335627":{"kind":"REWORK","_number":1,"created":"2026-09-07 13:28:34.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/1004420/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-tempest-plugin","ref":"refs/changes/20/1004420/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/20/1004420/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/20/1004420/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/20/1004420/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/20/1004420/1"}}},"commit":{"parents":[{"commit":"8b99294e3654420ad8001018959826916a7684db","subject":"Re-enable all jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/8b99294e3654420ad8001018959826916a7684db"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-07 13:28:26.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-07 13:28:26.000000000","tz":120},"subject":"Expect 404 return code for non-existent resource deletion","message":"Expect 404 return code for non-existent resource deletion\n\nWith API changes to properly reply with 404 when we attempt to delete\nsmth that does not exist, should not return 204, but 404.\nWe need to adjust tests to expect for such reply.\n\nNeeded-By: https://review.opendev.org/c/openstack/freezer-api/+/1004413\nChange-Id: Ieeb625f895b3f73f4a0b7d38c231f7391dcd4c9e\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/b2a919410233e389573da1c41ead0c0088335627"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/b2a919410233e389573da1c41ead0c0088335627"}]},"branch":"refs/heads/master"},"e5d62a3a7abebd1569e06d27a50879d78f3ea218":{"kind":"REWORK","_number":2,"created":"2026-09-08 13:54:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/1004420/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-tempest-plugin","ref":"refs/changes/20/1004420/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/20/1004420/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/20/1004420/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/20/1004420/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/20/1004420/2"}}},"commit":{"parents":[{"commit":"8b99294e3654420ad8001018959826916a7684db","subject":"Re-enable all jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/8b99294e3654420ad8001018959826916a7684db"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-07 13:28:26.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-08 13:54:07.000000000","tz":0},"subject":"Expect 404 return code for non-existent resource deletion","message":"Expect 404 return code for non-existent resource deletion\n\nWith API changes to properly reply with 404 when we attempt to delete\nsmth that does not exist, should not return 204, but 404.\nWe need to adjust tests to expect for such reply.\n\nNeeded-By: https://review.opendev.org/c/openstack/freezer-api/+/1004413\nChange-Id: Ieeb625f895b3f73f4a0b7d38c231f7391dcd4c9e\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/e5d62a3a7abebd1569e06d27a50879d78f3ea218"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/e5d62a3a7abebd1569e06d27a50879d78f3ea218"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
