)]}'
{"id":"openstack%2Ffreezer-tempest-plugin~1004645","triplet_id":"openstack%2Ffreezer-tempest-plugin~master~Ib9d363eea1171cc40df88e04ae458360536686ab","project":"openstack/freezer-tempest-plugin","branch":"master","topic":"freezer/backups_update","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-15 20:42:38.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Ib9d363eea1171cc40df88e04ae458360536686ab","subject":"Return CI jobs back","status":"MERGED","created":"2026-09-08 13:55:47.000000000","updated":"2026-09-15 22:03:33.000000000","submitted":"2026-09-15 22:03:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004645-freezer/backups_update","meta_rev_id":"aec0abaae2f4c4a40ac684faafd1818f62c28222","_number":1004645,"virtual_id_number":1004645,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-15 22:03:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"all":[{"value":1,"date":"2026-09-10 12:21:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},{"value":2,"date":"2026-09-09 08:12:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},{"value":1,"date":"2026-09-09 08:12:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-08 14:43:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-10 12:21:57.000000000","updated_by":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"reviewer":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"state":"REVIEWER"}],"messages":[{"id":"24f7b255fe05de7f9fabf2d50b5b73a938336608","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-08 13:55:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75ce22faac08bd626ace9f8e14e3276de23865d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 14:43:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a59756a3bcb64a16817407074f84a773\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4633ddc736d74188b9f9e9d1913438e2 : SUCCESS in 2m 52s\n- freezer-tempest-basic https://zuul.opendev.org/t/openstack/build/060aaf35ec4046609b99eb038f7d9a15 : SUCCESS in 39m 12s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/802fbbfebfcd4860a641a2e67299bacd : SUCCESS in 37m 08s\n- freezer-tempest-client https://zuul.opendev.org/t/openstack/build/571fc864afb0492bb828a189606c1a3d : SUCCESS in 40m 30s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/fdb7cd6c65ab4f0eaaffd9f9025064ad : SUCCESS in 40m 26s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/71bc2d3851fe4e6eb03ba6c75bfa6b68 : SUCCESS in 39m 02s","accounts_in_message":[],"_revision_number":1},{"id":"451c1f82f0a443ada148e333a4840cb640dfa68e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-09 08:12:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2109c292f57465bbb8af62bd9edc9790c7b4d56a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 08:12:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a48366d36f1427d901d9b754c88f29a\n\n\nWarning:\n  Change 1004413,2 in project openstack/freezer-api does not share a change queue with 1004645 in project openstack/freezer-tempest-plugin","accounts_in_message":[],"_revision_number":1},{"id":"0820b42a78598269b98683b8856d106c3c110328","author":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"date":"2026-09-10 12:21:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0755e6d39ecd323a79f29c317ede2737e00034a2","author":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"date":"2026-09-10 12:22:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0832a14b0d9ef2fa0e187e611e4008c3a5c7683e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-15 20:42:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83198fe68ea63f0c8f52f7ac7627340616f71dff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 21:19:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/395c6ff17c074b55a0f5627168b4e643\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/440a9b594bb040beb55409292f9f9e55 : SUCCESS in 2m 37s\n- freezer-tempest-basic https://zuul.opendev.org/t/openstack/build/e753a3bd3d31452fadb14701e92bf878 : SUCCESS in 34m 16s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94ea903770e24d76ac8c26e8e6f060b1 : SUCCESS in 27m 06s\n- freezer-tempest-client https://zuul.opendev.org/t/openstack/build/5e8a2f0bde9e45c797e83fc1dcfc8fdf : SUCCESS in 21m 50s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/4f2978c003494a619e614792a639799a : SUCCESS in 30m 11s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/8f87c77033cb4ab589480559e3fe3019 : SUCCESS in 24m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4143d3bd52511d25c4ca90eba10cd2f3345e80bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 21:20:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"44ec76a08d305717a5fb17660e810edd563be8d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:03:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a91ff0a2f32f4f8e8b5e3eb525b69762\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fae26c2dada24a36b403f77393b4ed47 : SUCCESS in 2m 47s\n- freezer-tempest-basic https://zuul.opendev.org/t/openstack/build/d146781930044740a6e57fc3d38a72fa : SUCCESS in 35m 37s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76cdf098b9674e8ca3c1c6c1d7378d74 : SUCCESS in 28m 36s\n- freezer-tempest-client https://zuul.opendev.org/t/openstack/build/451baf07a6d546d08f17086a96961d83 : SUCCESS in 20m 07s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/08173e32704a4015a1e9b075c3057eb6 : SUCCESS in 32m 47s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/a0f36b61c2da4e50ac3bedec6e720b3b : SUCCESS in 38m 01s","accounts_in_message":[],"_revision_number":1},{"id":"aec0abaae2f4c4a40ac684faafd1818f62c28222","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:03:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b77280d194b870110a95d06730e575b13b6c634c","revisions":{"b77280d194b870110a95d06730e575b13b6c634c":{"kind":"REWORK","_number":1,"created":"2026-09-08 13:55:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/45/1004645/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-tempest-plugin","ref":"refs/changes/45/1004645/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/45/1004645/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/45/1004645/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/45/1004645/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/45/1004645/1"}}},"commit":{"parents":[{"commit":"e5d62a3a7abebd1569e06d27a50879d78f3ea218","subject":"Expect 404 return code for non-existent resource deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/e5d62a3a7abebd1569e06d27a50879d78f3ea218"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-08 13:55:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-08 13:55:35.000000000","tz":120},"subject":"Return CI jobs back","message":"Return CI jobs back\n\nDepends-On: https://review.opendev.org/c/openstack/freezer-api/+/1004413\nChange-Id: Ib9d363eea1171cc40df88e04ae458360536686ab\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/b77280d194b870110a95d06730e575b13b6c634c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/b77280d194b870110a95d06730e575b13b6c634c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
