)]}'
{"id":"openstack%2Ffreezer-tempest-plugin~1005706","triplet_id":"openstack%2Ffreezer-tempest-plugin~master~Ibbafddb765216b443a3c0d20adf7d17348105105","project":"openstack/freezer-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ibbafddb765216b443a3c0d20adf7d17348105105","subject":"Re-enable the freezer-api tempest jobs","status":"ABANDONED","created":"2026-09-15 10:12:41.000000000","updated":"2026-09-17 08:13:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e94925447ff3a031ce6d36bb98e63de81a6ddf21","_number":1005706,"virtual_id_number":1005706,"owner":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-15 11:03:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-15 11:03:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6ee24ddda79bb0039e0adc0cd0f7c4d9dfd1489f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"date":"2026-09-15 10:12:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2666e439ea2b318c87c4aa99f903285dec339d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 11:03:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f731d0597e7c4cc9a0873c6ee9e6d6ae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cd914326e7e44bbb0f94ccc24200238 : SUCCESS in 1m 48s\n- freezer-tempest-basic https://zuul.opendev.org/t/openstack/build/b38bf3893b0c4ab4881b9c5079f58742 : SUCCESS in 35m 28s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3261e4586bd3495c94e7a8762b2ae490 : SUCCESS in 15m 57s\n- freezer-tempest-client https://zuul.opendev.org/t/openstack/build/d439dd5caca74c5292bdb81e58234113 : SUCCESS in 15m 51s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/0dd4d4dc75834eeb91ea4e9572974dba : SUCCESS in 22m 18s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/b09a8d780abc46f88443fdd1f8890b56 : SUCCESS in 27m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e94925447ff3a031ce6d36bb98e63de81a6ddf21","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-17 08:13:54.000000000","message":"Abandoned\n\nhas been abandoned by https://review.opendev.org/c/openstack/freezer-tempest-plugin/+/1004645","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"72f0618bfb97238176f2a52072c2a61ce20057f5","revisions":{"72f0618bfb97238176f2a52072c2a61ce20057f5":{"kind":"REWORK","_number":1,"created":"2026-09-15 10:12:41.000000000","uploader":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"ref":"refs/changes/06/1005706/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-tempest-plugin","ref":"refs/changes/06/1005706/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/06/1005706/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/06/1005706/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/06/1005706/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/06/1005706/1"}}},"commit":{"parents":[{"commit":"e5d62a3a7abebd1569e06d27a50879d78f3ea218","subject":"Expect 404 return code for non-existent resource deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/e5d62a3a7abebd1569e06d27a50879d78f3ea218"}]}],"author":{"name":"Alicja Filipek","email":"alicja.filipek@cleura.com","date":"2026-09-15 10:01:23.000000000","tz":120},"committer":{"name":"Alicja Filipek","email":"alicja.filipek@cleura.com","date":"2026-09-15 10:03:46.000000000","tz":120},"subject":"Re-enable the freezer-api tempest jobs","message":"Re-enable the freezer-api tempest jobs\n\nFreezer-tempest-basic and freezer-tempest-ipv6 were\ncommented out to enable the corresponding fix on freezer-api side.\nThe patch https://review.opendev.org/c/openstack/freezer-api/+/1004413\nhas already been merged.\n\nChange-Id: Ibbafddb765216b443a3c0d20adf7d17348105105\nSigned-off-by: Alicja Filipek \u003calicja.filipek@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/72f0618bfb97238176f2a52072c2a61ce20057f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/72f0618bfb97238176f2a52072c2a61ce20057f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
