)]}'
{"id":"openstack%2Ffreezer-tempest-plugin~682742","triplet_id":"openstack%2Ffreezer-tempest-plugin~master~Iad288fe17db561cf266ae3db5a66ea250fbace33","project":"openstack/freezer-tempest-plugin","branch":"master","topic":"ipv6-only-deployment-and-testing","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-11-27 17:04:28.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-11-27 17:47:46.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-11-27 17:47:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad288fe17db561cf266ae3db5a66ea250fbace33","subject":"[train][goal] Define new \u0027freezer-tempest-ipv6-only\u0027 job","status":"MERGED","created":"2019-09-17 20:25:46.000000000","updated":"2025-11-29 11:26:02.000000000","submitted":"2025-11-27 17:47:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682742-ipv6-only-deployment-and-testing","meta_rev_id":"6e4ab55da3eabc622a89ed6aec7bbe5374ddcae1","_number":682742,"virtual_id_number":682742,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-27 17:47:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2025-01-10 16:36:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2025-11-24 08:36:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2025-11-27 17:04:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-15 02:22:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-10 16:36:04.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-11-24 08:36:24.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-11-24 08:36:36.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"292af47216f8579f3e5e909b998daf41379d1f79","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-17 20:25:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e03babce1e77bd7d9670f797aca572afb26559a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-17 20:31:35.000000000","message":"Topic set to ipv6-only-deployment-and-testing","accounts_in_message":[],"_revision_number":1},{"id":"b066d6707fcfeb23278cf3b32370808abe1b72a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 21:13:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5be331b5d8814f8c91c765a82050ef3c : SUCCESS in 4m 33s\n- freezer-tempest-basic https://zuul.opendev.org/t/openstack/build/63ebff4593eb45c08a6478a2be3dd3b3 : SUCCESS in 33m 53s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c544baecf7f342b4962581f058702b77 : FAILURE in 34m 43s\n- freezer-tempest-client https://zuul.opendev.org/t/openstack/build/e85f6f7fd7724bd4825dea446b865fc1 : SUCCESS in 40m 56s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/04c72d6891dd4ee09da3ae88075ac872 : SUCCESS in 38m 19s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/f30266463a53477eab451fd9933ffe89 : SUCCESS in 34m 42s\n- freezer-tempest-basic-sql https://zuul.opendev.org/t/openstack/build/5fbc7e3033904d8587f3b6fec03de39a : SUCCESS in 36m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f92db95337760ecbdee9fd53cc94623a0f1c4274","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-29 21:22:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2ea98ddd8e8c7447ce7b68861326bb1048dad36a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 22:55:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e357f305a8641f68fc66efbd4fe6520 : SUCCESS in 3m 30s\n- freezer-tempest-basic https://zuul.opendev.org/t/openstack/build/1d36bc624aac422c9a0bc30f975f979c : SUCCESS in 35m 03s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/264d79b9e8ee4761bf540ff553e2b02b : FAILURE in 33m 55s\n- freezer-tempest-client https://zuul.opendev.org/t/openstack/build/f7c138733df64f4ca6fe05b926099fa6 : SUCCESS in 31m 01s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/ab4108c7fe144b548a9d515400ee8e11 : SUCCESS in 33m 07s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/357ab448d2504760b24706167380f47e : SUCCESS in 32m 04s\n- freezer-tempest-basic-sql https://zuul.opendev.org/t/openstack/build/fd799ef0fa7c4dc6ae49ee5484c56287 : SUCCESS in 32m 28s","accounts_in_message":[],"_revision_number":1},{"id":"1afcd549ee2cf3468628f126e9e715145cc38ea9","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-11-15 01:38:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"8f1636e609d48a26a3c437e51ef8987770abbbf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 02:22:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67c6f26d4705481c9646b8e4eaa7de7e : SUCCESS in 2m 38s\n- freezer-tempest-basic https://zuul.opendev.org/t/openstack/build/fcec0818ce264ff3be0100cdc2cfb10a : SUCCESS in 34m 54s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fbd7eb1452454b3b97102b34f95cc783 : FAILURE in 34m 16s\n- freezer-tempest-client https://zuul.opendev.org/t/openstack/build/f0f812b64cb0434e939c4630bfce9e3e : SUCCESS in 38m 21s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/93e8c8b38eb6488b997807771c47569b : SUCCESS in 36m 57s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/819bf3743d6346679437a667c453a846 : SUCCESS in 33m 42s\n- freezer-tempest-basic-sql https://zuul.opendev.org/t/openstack/build/aa08965141c84d198b138916b08d773d : SUCCESS in 31m 14s","accounts_in_message":[],"_revision_number":2},{"id":"052b04363591928a910c7ce5a8cc0f832b8047ef","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-02 16:56:40.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nThe following files contain Git conflicts:\n* .zuul.yaml\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c279e8bc618d931af75c494babae985083f14fd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 16:57:08.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/freezer-tempest-plugin on branch master.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/freezer-tempest-plugin/.zuul.yaml@master\", line 21, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/freezer-tempest-plugin/.zuul.yaml@master\", line 23, column 1","accounts_in_message":[],"_revision_number":3},{"id":"793caf6fe6e66e47eb108a2763919903423be672","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-02 17:00:02.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2c1d091713f1f9af37e5304f1ba9536c2aac0349","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 17:00:26.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/freezer-tempest-plugin on branch master.  The\nproblem was:\n\n  extra keys not allowed @ data[\u0027gate\u0027][\u0027queue\u0027]\n\nThe problem appears in the the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - tempest-plugin-jobs\n      check:\n        jobs:\n          - freezer-tempest-basic\n          - freezer-tempest-ipv6-only\n          - freezer-tempest-client\n  ...\n\n  in \"openstack/freezer-tempest-plugin/.zuul.yaml@master\", line 1","accounts_in_message":[],"_revision_number":4},{"id":"1bab12871bb43b701b072830aef5b2f41f4d310a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-12-03 19:04:13.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"05152c19c7759fc679f6a1324e37f068788d144e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 19:04:35.000000000","message":"Patch Set 5: Verified-1\n\n(7 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/freezer-tempest-plugin on branch master.  The\nproblem was:\n\n  The project \"openstack/devstack-gate\" was not found.  All projects\n  referenced within a Zuul configuration must first be added to the main\n  configuration file by the Zuul administrator.\n\nThe problem appears in the the \"freezer-tempest-basic\" job stanza:\n\n  job:\n      name: freezer-tempest-basic\n      parent: devstack-tempest\n      # TODO(gmann): Remove the below nodeset setting to Bionic once\n      # bug https://bugs.launchpad.net/freezer/+bug/1896648 is fixed.\n      # Once nodeset is removed form here then devstack-tempest job\n      # will automatically run this job on Ubuntu Focal nodeset from\n      # Victoria gate onwards.\n      # nodeset: openstack-single-node-bionic\n  ...\n\n  in \"openstack/freezer-tempest-plugin/.zuul.yaml@master\", line 22","accounts_in_message":[],"_revision_number":5},{"id":"799c7c33841ba33a1b47f8ca0c1c9d0cd6e5b74c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-12-03 21:00:02.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"a2bea099f859b04ffc26f795bfb8d5c8286b9029","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 21:50:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/367fb9bd4b56430cb9638080e793578d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcf70e5445e24e3fabb30497229d5b73 : SUCCESS in 2m 23s\n- freezer-tempest-basic https://zuul.opendev.org/t/openstack/build/68c1c4d22eac4c5c989114d4b8e84c6f : SUCCESS in 39m 12s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64750acbeadf4e939d36930ec22ddc6e : SUCCESS in 24m 17s\n- freezer-tempest-client https://zuul.opendev.org/t/openstack/build/e6049eb9b9714168acb4426fe335a5e6 : SUCCESS in 43m 49s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/b62ac8f917234a3c80f3429e2e7b3ca5 : SUCCESS in 23m 30s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/1955f77af11f49e28b19d1edbee8c387 : SUCCESS in 20m 59s","accounts_in_message":[],"_revision_number":6},{"id":"6950a1c4e1f91895173708795ac8fff557b21edd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-01-10 16:36:04.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"cce871d1b0e2d835cb0d45e14f433af81fbe327b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-24 08:36:24.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d2a59d139c2b311b4f315c3037ca8995ba9932fb","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-24 08:36:36.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"ae371de4cb18296096774b46bcf4d16e261d479c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-27 17:03:09.000000000","message":"Patch Set 6:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":6},{"id":"53ac0097c5ecc3c04f515298d1fa03f7a0e0a0e6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-11-27 17:04:28.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"87cb07bdf3f14818146a225281f75d48452dad53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 17:05:12.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b2a3d8d0a66587b8a51d8a1c5d5a3d3a77b80f18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 17:47:45.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6288d5d1169a4187a3315ab6b405b02a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53de288051104277afdce9a2ead5149e : SUCCESS in 1m 36s\n- freezer-tempest-basic https://zuul.opendev.org/t/openstack/build/cd44b99ce5214202b7bdf94f738dd735 : SUCCESS in 34m 11s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d94a4b4f99dd46bc9262e7f0226e24b2 : SUCCESS in 15m 49s\n- freezer-tempest-client https://zuul.opendev.org/t/openstack/build/6e7cddaf75684bceb1584c540bb5f5d8 : SUCCESS in 37m 45s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/aa4ba8de972448cea4004c4868a50dbc : SUCCESS in 35m 16s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/f8a0bc06980d42a58dde4d1f68964d69 : SUCCESS in 15m 51s","accounts_in_message":[],"_revision_number":6},{"id":"aeb48b6f61105ed2a853bc04ea5b90a1637d79d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 17:47:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"6e4ab55da3eabc622a89ed6aec7bbe5374ddcae1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-29 11:26:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"762a4d657525a5fd41f33b3e60afdd3c1d6e69b7","revisions":{"1a70f9f8b7c829ce1f522ab04a5863a4f24c2712":{"kind":"REWORK","_number":1,"created":"2019-09-17 20:25:46.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/42/682742/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-tempest-plugin","ref":"refs/changes/42/682742/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/1"}}},"commit":{"parents":[{"commit":"ca93c45669d30c8e672fb69f21dc1dfbcd3b9e8d","subject":"Fix ci fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/ca93c45669d30c8e672fb69f21dc1dfbcd3b9e8d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 20:24:11.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 20:25:44.000000000","tz":0},"subject":"[train][goal] Define new \u0027freezer-tempest-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027freezer-tempest-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027freezer-tempest-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027freezer-tempest-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35890\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Iad288fe17db561cf266ae3db5a66ea250fbace33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/1a70f9f8b7c829ce1f522ab04a5863a4f24c2712"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/1a70f9f8b7c829ce1f522ab04a5863a4f24c2712"}]},"branch":"refs/heads/master"},"63ee0464792b9c729c6adbe603fc66bd74bce4d4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-11-15 01:38:10.000000000","uploader":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"ref":"refs/changes/42/682742/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-tempest-plugin","ref":"refs/changes/42/682742/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/2"}}},"commit":{"parents":[{"commit":"79d9ad5ed11289ba588261b1e9992bfed1a9a254","subject":"Merge \"Sync Sphinx requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/79d9ad5ed11289ba588261b1e9992bfed1a9a254"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 20:24:11.000000000","tz":0},"committer":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2019-11-15 01:38:10.000000000","tz":0},"subject":"[train][goal] Define new \u0027freezer-tempest-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027freezer-tempest-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027freezer-tempest-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027freezer-tempest-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35890\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Iad288fe17db561cf266ae3db5a66ea250fbace33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/63ee0464792b9c729c6adbe603fc66bd74bce4d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/63ee0464792b9c729c6adbe603fc66bd74bce4d4"}]},"branch":"refs/heads/master"},"b575f76b59143538bec2d63ae24f254d0a42532f":{"kind":"REWORK","_number":3,"created":"2024-05-02 16:56:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/42/682742/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-tempest-plugin","ref":"refs/changes/42/682742/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/3"}}},"commit":{"parents":[{"commit":"f90cff9a91270cfcc1fd9fb7bef6ba934c2cf0eb","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/f90cff9a91270cfcc1fd9fb7bef6ba934c2cf0eb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 20:24:11.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-05-02 16:56:40.000000000","tz":0},"subject":"[train][goal] Define new \u0027freezer-tempest-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027freezer-tempest-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027freezer-tempest-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027freezer-tempest-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35890\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Iad288fe17db561cf266ae3db5a66ea250fbace33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/b575f76b59143538bec2d63ae24f254d0a42532f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/b575f76b59143538bec2d63ae24f254d0a42532f"}]},"branch":"refs/heads/master","description":"Rebase"},"85f93c1f4482f30cb722eeb05dc7f1c34fadf695":{"kind":"REWORK","_number":4,"created":"2024-05-02 17:00:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/42/682742/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-tempest-plugin","ref":"refs/changes/42/682742/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/4"}}},"commit":{"parents":[{"commit":"f90cff9a91270cfcc1fd9fb7bef6ba934c2cf0eb","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/f90cff9a91270cfcc1fd9fb7bef6ba934c2cf0eb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 20:24:11.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-02 17:00:01.000000000","tz":120},"subject":"[train][goal] Define new \u0027freezer-tempest-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027freezer-tempest-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027freezer-tempest-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027freezer-tempest-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35890\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Iad288fe17db561cf266ae3db5a66ea250fbace33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/85f93c1f4482f30cb722eeb05dc7f1c34fadf695"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/85f93c1f4482f30cb722eeb05dc7f1c34fadf695"}]},"branch":"refs/heads/master"},"87b1d3cc6863aa0a808dd4bb274b55f730366c0e":{"kind":"REWORK","_number":5,"created":"2024-12-03 19:04:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/42/682742/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-tempest-plugin","ref":"refs/changes/42/682742/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/5"}}},"commit":{"parents":[{"commit":"549885a16a569065f07d15a877760ebc733da92c","subject":"cli without project_name argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/549885a16a569065f07d15a877760ebc733da92c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 20:24:11.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-03 19:04:07.000000000","tz":60},"subject":"[train][goal] Define new \u0027freezer-tempest-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027freezer-tempest-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027freezer-tempest-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027freezer-tempest-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35890\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Iad288fe17db561cf266ae3db5a66ea250fbace33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/87b1d3cc6863aa0a808dd4bb274b55f730366c0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/87b1d3cc6863aa0a808dd4bb274b55f730366c0e"}]},"branch":"refs/heads/master"},"762a4d657525a5fd41f33b3e60afdd3c1d6e69b7":{"kind":"REWORK","_number":6,"created":"2024-12-03 21:00:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/42/682742/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-tempest-plugin","ref":"refs/changes/42/682742/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/42/682742/6"}}},"commit":{"parents":[{"commit":"549885a16a569065f07d15a877760ebc733da92c","subject":"cli without project_name argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/549885a16a569065f07d15a877760ebc733da92c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 20:24:11.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-03 21:00:02.000000000","tz":0},"subject":"[train][goal] Define new \u0027freezer-tempest-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027freezer-tempest-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027freezer-tempest-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027freezer-tempest-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35890\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Iad288fe17db561cf266ae3db5a66ea250fbace33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/762a4d657525a5fd41f33b3e60afdd3c1d6e69b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/762a4d657525a5fd41f33b3e60afdd3c1d6e69b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
