)]}'
{"id":"openstack%2Ffreezer-tempest-plugin~725703","triplet_id":"openstack%2Ffreezer-tempest-plugin~master~I7d1db17f492d873ee7408286f8b83c325576012f","project":"openstack/freezer-tempest-plugin","branch":"master","topic":"stable-testing-tempest-plugins","hashtags":[],"change_id":"I7d1db17f492d873ee7408286f8b83c325576012f","subject":"Add stable branch jobs on the plugins master gate","status":"MERGED","created":"2020-05-05 22:17:01.000000000","updated":"2020-05-23 07:23:28.000000000","submitted":"2020-05-23 07:23:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725703-1590218608294-fdce670a","meta_rev_id":"d32e2048edb882903a9cdbc1cbe6be15527b3adb","_number":725703,"virtual_id_number":725703,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-23 07:23:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-19 01:44:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-23 06:33:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-23 06:33:00.000000000","updated_by":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"reviewer":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"state":"REVIEWER"},{"updated":"2020-05-23 07:23:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12f8a308482b854608a4c42bf10530351b4b4c38","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-05 22:17:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e8a38d8ec09fc81e502fee6ae509fef9f8848b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 23:10:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94eacbf8d5844f7ea0d58597a0f877b0 : SUCCESS in 4m 05s\n- freezer-tempest-basic https://zuul.opendev.org/t/openstack/build/a1aaba9e4285482687ea9d1e6b6c88b4 : SUCCESS in 32m 15s\n- freezer-tempest-client https://zuul.opendev.org/t/openstack/build/7498e15f183e4e6e80ce7032d680583a : SUCCESS in 50m 53s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/3c00d35c2a234446a0e5edb946c857df : SUCCESS in 44m 59s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/44b0d680081245e0861d447a68330a18 : FAILURE in 33m 19s\n- freezer-tempest-basic-sql https://zuul.opendev.org/t/openstack/build/ba1e4872e5f2441b884d9e50692630b5 : SUCCESS in 38m 02s\n- freezer-tempest-basic-ussuri https://zuul.opendev.org/t/openstack/build/022d18feb12a41c8bbc335e305e1a531 : SUCCESS in 45m 20s\n- freezer-tempest-basic-train https://zuul.opendev.org/t/openstack/build/f972fd8262e745c795af7a29a338e239 : SUCCESS in 32m 37s\n- freezer-tempest-basic-stein https://zuul.opendev.org/t/openstack/build/a262893fa26044f0befc7bf0da76587a : SUCCESS in 42m 04s","accounts_in_message":[],"_revision_number":1},{"id":"3d47f067d3f41da2864c2659e1dec6a446f97eaa","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-06 00:17:46.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2f30fb8a39e575898830b22360fee7cf176d0ea3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 01:07:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1f3900a878f47c28209814dd54f0db3 : SUCCESS in 3m 38s\n- freezer-tempest-basic https://zuul.opendev.org/t/openstack/build/b2366fb591f54970b479daf04d91365f : SUCCESS in 34m 35s\n- freezer-tempest-client https://zuul.opendev.org/t/openstack/build/d1d1095cc1504f37892f907d90d424a0 : SUCCESS in 38m 26s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/ef48bd1803514ef6bc81b94b19813349 : SUCCESS in 48m 37s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/ccc6f613bb1e40099c7c8520c4fae25b : SUCCESS in 32m 42s\n- freezer-tempest-basic-sql https://zuul.opendev.org/t/openstack/build/7532f7f561ce49cea0d74f4ee4b22974 : SUCCESS in 30m 26s\n- freezer-tempest-basic-ussuri https://zuul.opendev.org/t/openstack/build/155535e3337f4f0880f470a298b2e783 : SUCCESS in 32m 13s\n- freezer-tempest-basic-train https://zuul.opendev.org/t/openstack/build/f5b6850421f54aafa2889ffc62843a4e : SUCCESS in 35m 08s\n- freezer-tempest-basic-stein https://zuul.opendev.org/t/openstack/build/089e5dcdda224019af0e7834c4965c93 : SUCCESS in 35m 08s","accounts_in_message":[],"_revision_number":2},{"id":"3164547f7d14a9b25640bad6b17549c2e0eba78f","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-05-19 01:44:44.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM.\nThanks.","accounts_in_message":[],"_revision_number":2},{"id":"3395ada962d6201dc6ffd7c045ffcb4e1ccf714e","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-05-23 06:33:00.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f3c7e2347455dd64de393c5f744d8479ea942f8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-23 06:33:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6d14dc944f7ae39ea0114e5f64abe2ffd8f7cd95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-23 07:23:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"477770d3e78d390d6d065e8f2589203aa4f402ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-23 07:23:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7425b8d3154e401f8e5555ed69012c94 : SUCCESS in 5m 32s\n- freezer-tempest-basic https://zuul.opendev.org/t/openstack/build/8b1b03e2605b41c6ab8447c05907a5ba : SUCCESS in 39m 37s\n- freezer-tempest-client https://zuul.opendev.org/t/openstack/build/1f48bba1b6d84b8c8007b9754fe6b6a7 : SUCCESS in 39m 14s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/ac53bc0842494c628666119a972e8321 : SUCCESS in 38m 38s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/45e437abf9a047a8a83ee990ed4c01d7 : SUCCESS in 43m 43s\n- freezer-tempest-basic-sql https://zuul.opendev.org/t/openstack/build/d4485f11d0b1463cb7e7c61e48e60118 : SUCCESS in 35m 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3300282acd8b53be6e6800db1ed8e4147a2645ea","revisions":{"034ed3c1fe35359acae1dfbf6defcc950acc9063":{"kind":"REWORK","_number":1,"created":"2020-05-05 22:17:01.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/03/725703/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-tempest-plugin","ref":"refs/changes/03/725703/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/03/725703/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/03/725703/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/03/725703/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/03/725703/1"}}},"commit":{"parents":[{"commit":"f75709dda5281a1d95842cc1070d08a4127a67da","subject":"Merge \"Fix tests to use path_to_backup instead src_file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/f75709dda5281a1d95842cc1070d08a4127a67da"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-04 23:25:32.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-05 22:15:37.000000000","tz":-300},"subject":"Add stable branch jobs on the plugins master gate","message":"Add stable branch jobs on the plugins master gate\n\nSupported stable branches in maintenance state\n(stein, train, ussuri) use the Tempest and plugins master\nversion for their testing. To make sure master version (new\ntests or any changes) is compatible with stable branch testing,\nwe need to add the stable branches job on the plugins master gate.\nTempest and many other plugins are already testing the stable\nbranches in same way[1].\n\nUssuri onwards, Tempest or plugins are py3 only, so we are\nrunning only python3 version of stable jobs\n\nTempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\n[1] https://opendev.org/openstack/tempest/src/commit/e8f1876aa6772077f85f380677b30251c2454505/.zuul.yaml#L646\n\nChange-Id: I7d1db17f492d873ee7408286f8b83c325576012f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/034ed3c1fe35359acae1dfbf6defcc950acc9063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/034ed3c1fe35359acae1dfbf6defcc950acc9063"}]},"branch":"refs/heads/master"},"3300282acd8b53be6e6800db1ed8e4147a2645ea":{"kind":"REWORK","_number":2,"created":"2020-05-06 00:17:46.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/03/725703/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-tempest-plugin","ref":"refs/changes/03/725703/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/03/725703/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/03/725703/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/03/725703/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/03/725703/2"}}},"commit":{"parents":[{"commit":"f75709dda5281a1d95842cc1070d08a4127a67da","subject":"Merge \"Fix tests to use path_to_backup instead src_file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/f75709dda5281a1d95842cc1070d08a4127a67da"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-04 23:25:32.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-06 00:17:44.000000000","tz":0},"subject":"Add stable branch jobs on the plugins master gate","message":"Add stable branch jobs on the plugins master gate\n\nSupported stable branches in maintenance state\n(stein, train, ussuri) use the Tempest and plugins master\nversion for their testing. To make sure master version (new\ntests or any changes) is compatible with stable branch testing,\nwe need to add the stable branches job on the plugins master gate.\nTempest and many other plugins are already testing the stable\nbranches in same way[1].\n\nUssuri onwards, Tempest or plugins are py3 only, so we are\nrunning only python3 version of stable jobs\n\nTempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\n[1] https://opendev.org/openstack/tempest/src/commit/e8f1876aa6772077f85f380677b30251c2454505/.zuul.yaml#L646\n\nChange-Id: I7d1db17f492d873ee7408286f8b83c325576012f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/3300282acd8b53be6e6800db1ed8e4147a2645ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/3300282acd8b53be6e6800db1ed8e4147a2645ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
