)]}'
{"id":"openstack%2Ffreezer-tempest-plugin~991704","triplet_id":"openstack%2Ffreezer-tempest-plugin~master~I69375bb62cb9bbd2763e5885f822e11c6ca2b6a0","project":"openstack/freezer-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-06-05 07:30:31.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I69375bb62cb9bbd2763e5885f822e11c6ca2b6a0","subject":"Remove user_name from expected return","status":"MERGED","created":"2026-06-04 17:59:09.000000000","updated":"2026-06-05 08:49:32.000000000","submitted":"2026-06-05 08:49:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991704","meta_rev_id":"68d497b7470b500ec03bc9a90dcfc76e4d07868a","_number":991704,"virtual_id_number":991704,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-05 08:49:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-06-05 07:08:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-06-05 07:08:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-04 18:47:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6f11d21fe2c26a48d1e89c651394740d1054b8a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-04 17:59:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"527bb0c05254aad82c07f7ee17cf3679b14d8575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 18:47:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a717385f342438baae4353d7bddb0d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02c784b489134f09ac563e09f79b20ef : SUCCESS in 2m 24s\n- freezer-tempest-basic https://zuul.opendev.org/t/openstack/build/4858fbc2b0ea493a9c278bf91f135ea9 : SUCCESS in 38m 18s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d3fae040a0a40719da9c20f8594785d : SUCCESS in 35m 41s\n- freezer-tempest-client https://zuul.opendev.org/t/openstack/build/e54d9acba9024852988db8f9aae81f3e : SUCCESS in 33m 59s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/80f1620f40694e9caa14db0eeed09bc8 : SUCCESS in 37m 43s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/b189de21cea1423fa297fab35cae4b58 : SUCCESS in 37m 24s","accounts_in_message":[],"_revision_number":1},{"id":"36363d71797f57c40b09d5a1920595916e8f9113","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-05 07:08:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8cfbbac98f87ac78e8a4f97818eab32df0d19d5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 07:08:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a65950cf34d4dabad3934b6888d6dc2\n\n\nWarning:\n  Change 990963,3 in project openstack/freezer-api does not share a change queue with 991704 in project openstack/freezer-tempest-plugin","accounts_in_message":[],"_revision_number":1},{"id":"623a5ae8ad906316a7fb8f75722ce03f30473490","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-05 07:30:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f2221026b8ceb6ab6cb5f0f2f463f0aac556514","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 08:06:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3541d4ffbce343c0b0fb80508f15e0eb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5b34096946b4cf69c96da3f76bd8864 : SUCCESS in 1m 55s\n- freezer-tempest-basic https://zuul.opendev.org/t/openstack/build/23714fa52bd148799cd3e9cd6e85e064 : SUCCESS in 15m 57s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56ddc6fd81d648b499f6c8a334a511d9 : SUCCESS in 15m 40s\n- freezer-tempest-client https://zuul.opendev.org/t/openstack/build/270bc52a04cb4decaf70562fef521d7a : SUCCESS in 18m 27s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/2e602084c4c14603a429328efa8a37ca : SUCCESS in 26m 57s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/33d95baa3e7a4849b756d49d623cd13d : SUCCESS in 33m 45s","accounts_in_message":[],"_revision_number":1},{"id":"213f667f89e8dca591a9b0d2af25bb2ef1b79f2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 08:06:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b1d34acd214185f5cb54b20790f8236e68745525","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 08:49:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fc164ab72e74a5989b09965564953c5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f3fea2725fe42baab0547dbfce68501 : SUCCESS in 3m 30s\n- freezer-tempest-basic https://zuul.opendev.org/t/openstack/build/94d6127612e44da7bc1dc3d1959eb45c : SUCCESS in 15m 34s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1e3040149cf4422a6e8b7aad998046e : SUCCESS in 38m 03s\n- freezer-tempest-client https://zuul.opendev.org/t/openstack/build/86eb21e80b6c49f4a70e94a9eff064f6 : SUCCESS in 15m 29s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/64ff2ea06a8543cc8345706a637baffd : SUCCESS in 15m 02s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/60b631e5bdfb4a1cb165e1b2c52d4db0 : SUCCESS in 16m 31s","accounts_in_message":[],"_revision_number":1},{"id":"68d497b7470b500ec03bc9a90dcfc76e4d07868a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 08:49:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ccc67d8c9bfc14edc11a64bb96190ffa71ec09d","revisions":{"9ccc67d8c9bfc14edc11a64bb96190ffa71ec09d":{"kind":"REWORK","_number":1,"created":"2026-06-04 17:59:09.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/04/991704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-tempest-plugin","ref":"refs/changes/04/991704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/04/991704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/04/991704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/04/991704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-tempest-plugin refs/changes/04/991704/1"}}},"commit":{"parents":[{"commit":"88897ddbf4813ee3fc08be1e80675d3acdf58218","subject":"Merge \"Use unique job id for tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/88897ddbf4813ee3fc08be1e80675d3acdf58218"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-04 17:59:01.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-04 17:59:01.000000000","tz":120},"subject":"Remove user_name from expected return","message":"Remove user_name from expected return\n\nDepends-On: https://review.opendev.org/c/openstack/freezer-api/+/990963\nChange-Id: I69375bb62cb9bbd2763e5885f822e11c6ca2b6a0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/9ccc67d8c9bfc14edc11a64bb96190ffa71ec09d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-tempest-plugin/commit/9ccc67d8c9bfc14edc11a64bb96190ffa71ec09d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
