)]}'
{"id":"openstack%2Ffreezer-web-ui~694489","triplet_id":"openstack%2Ffreezer-web-ui~master~Ibb7a5ad85343d0ffe5c50d6c72865bce6657ea36","project":"openstack/freezer-web-ui","branch":"master","hashtags":[],"change_id":"Ibb7a5ad85343d0ffe5c50d6c72865bce6657ea36","subject":"Add  py37 and remove py35 in envlst","status":"MERGED","created":"2019-11-15 09:23:20.000000000","updated":"2019-11-15 12:07:18.000000000","submitted":"2019-11-15 12:07:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694489-1573819638508-10f32a19","meta_rev_id":"d27311775e4483808c73da69b03c6695dae8655a","_number":694489,"virtual_id_number":694489,"owner":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-15 12:07:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-15 11:56:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-15 11:56:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-15 12:07:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"39d045765be0e1da67a1998f40477c27c9172846","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2019-11-15 09:23:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc1f51ba91adf021d06d167c3672fe6a96b71077","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2019-11-15 09:24:16.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"46773eb3d45c138b7dcce6751a48582e927f83ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 10:02:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/fed8c16a2b8f4d499fdcf7daa5123416 : SUCCESS in 4m 54s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/c679d0ea33f1483d801420f0b2ca8fc8 : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/303ad424613b40f985a17e576e0115ef : FAILURE in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9703880ac5124843afb33105a8eeda2f : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d28977c85c7b408eb816759b4415387d : SUCCESS in 5m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8f967d81644e454c868806cbea5b1544 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c31273d723534b94b0df68859d32fe9a : SUCCESS in 6m 13s\n- freezer-web-ui-ubuntu https://zuul.opendev.org/t/openstack/build/7d631c12d3b940e69dabaafdc1cf1975 : SUCCESS in 33m 12s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dbb3f9d151154720b2e82d63d87e449b : FAILURE in 4m 35s","accounts_in_message":[],"_revision_number":2},{"id":"4b1a2581620feec612446a51a068d8619aaf1c7f","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2019-11-15 10:25:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6ec1ae20a6a2b380f88c09d324f411691fb0f4b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 10:59:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9c18f59d6f4849348c516b0acafc958f : SUCCESS in 2m 21s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/05d27d0cad39405391fccc2e65635a92 : SUCCESS in 6m 20s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/0c93a3ca35c948d8b6caee7b5e46d31a : SUCCESS in 7m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4e09edd7ea745d0a5d692f44ce00579 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b87b0b01b2344ca9a29a36583ccfa076 : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cfae69f0abd54f919cf070a7ea4fb154 : SUCCESS in 3m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/768296ad6caa4da99361393143571f1c : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56a7522f80d14e38bc7264cc180aa6f1 : SUCCESS in 4m 40s\n- freezer-web-ui-ubuntu https://zuul.opendev.org/t/openstack/build/26d30c325c08499f9d0dce40986be024 : SUCCESS in 31m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b7b80325f6a04b17bbac6b4b5f4deac7 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":3},{"id":"a01e9d176315474b9f7252281d5a76da483283b5","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2019-11-15 11:56:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"98fb046304a9aeac67483b6bf9c06bfed63cfd3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 11:56:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"28d27d7471213edfe3bd8f60d3d67dad8d9980dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 12:07:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d27311775e4483808c73da69b03c6695dae8655a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 12:07:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1ca785f6e60a4510851557d00cd9b07f : SUCCESS in 3m 29s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/9d6a2cac749845b09463264fc239e7b6 : SUCCESS in 5m 39s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/e9184f6614004489b93b69931ce1a077 : SUCCESS in 5m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b03ca0656ab44ca912fa5449509c68e : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c22b83991cc64af891fc82236dd1b24b : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d077703bb124d2cbb4972769a6283f8 : SUCCESS in 4m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3d7b39b3dad6424ca0fecf95002cd088 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32086ffe41e14def924947ccda65721a : SUCCESS in 3m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/598711d894f34097aa2b41873bdeb859 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f894d59db0ce07e81d6f780a4090821f45b5b3d8","revisions":{"3c06c3bf39b70e34dd3296cee215bc24865b1348":{"kind":"REWORK","_number":1,"created":"2019-11-15 09:23:20.000000000","uploader":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"ref":"refs/changes/89/694489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/89/694489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/89/694489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/89/694489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/89/694489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/89/694489/1"}}},"commit":{"parents":[{"commit":"146a150bc1ed7bb28fb7db53afb8537ac2952e20","subject":"Use Horizon project template for django jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/146a150bc1ed7bb28fb7db53afb8537ac2952e20"}]}],"author":{"name":"caihui","email":"cai.hui@zte.com.cn","date":"2019-11-15 09:20:00.000000000","tz":-480},"committer":{"name":"caihui","email":"cai.hui@zte.com.cn","date":"2019-11-15 09:23:16.000000000","tz":-480},"subject":"Add  py37 and remove py35 in envlst","message":"Add  py37 and remove py35 in envlst\n\nUpdate tox.ini file, add py37 and remove py35 in envlst for UT.\n\nChange-Id: Ibb7a5ad85343d0ffe5c50d6c72865bce6657ea36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/3c06c3bf39b70e34dd3296cee215bc24865b1348"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/3c06c3bf39b70e34dd3296cee215bc24865b1348"}]},"branch":"refs/heads/master"},"a8f6365c1f234510aa86bcba30c6ce30e2f87a79":{"kind":"NO_CHANGE","_number":2,"created":"2019-11-15 09:24:16.000000000","uploader":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"ref":"refs/changes/89/694489/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/89/694489/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/89/694489/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/89/694489/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/89/694489/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/89/694489/2"}}},"commit":{"parents":[{"commit":"146a150bc1ed7bb28fb7db53afb8537ac2952e20","subject":"Use Horizon project template for django jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/146a150bc1ed7bb28fb7db53afb8537ac2952e20"}]}],"author":{"name":"caihui","email":"cai.hui@zte.com.cn","date":"2019-11-15 09:20:00.000000000","tz":-480},"committer":{"name":"caihui","email":"cai.hui@zte.com.cn","date":"2019-11-15 09:23:32.000000000","tz":-480},"subject":"Add  py37 and remove py35 in envlst","message":"Add  py37 and remove py35 in envlst\n\nUpdate tox.ini file, add py37 and remove py35 in envlst for UT.\n\nChange-Id: Ibb7a5ad85343d0ffe5c50d6c72865bce6657ea36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/a8f6365c1f234510aa86bcba30c6ce30e2f87a79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/a8f6365c1f234510aa86bcba30c6ce30e2f87a79"}]},"branch":"refs/heads/master"},"f894d59db0ce07e81d6f780a4090821f45b5b3d8":{"kind":"REWORK","_number":3,"created":"2019-11-15 10:25:13.000000000","uploader":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"ref":"refs/changes/89/694489/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/89/694489/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/89/694489/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/89/694489/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/89/694489/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/89/694489/3"}}},"commit":{"parents":[{"commit":"146a150bc1ed7bb28fb7db53afb8537ac2952e20","subject":"Use Horizon project template for django jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/146a150bc1ed7bb28fb7db53afb8537ac2952e20"}]}],"author":{"name":"caihui","email":"cai.hui@zte.com.cn","date":"2019-11-15 09:20:00.000000000","tz":-480},"committer":{"name":"caihui","email":"cai.hui@zte.com.cn","date":"2019-11-15 10:21:33.000000000","tz":-480},"subject":"Add  py37 and remove py35 in envlst","message":"Add  py37 and remove py35 in envlst\n\n1. Update tox.ini file, add py37 and remove py35 in envlst for UT.\n2. Upgrade pylint,astroid,keystoneauth1  version\n\nChange-Id: Ibb7a5ad85343d0ffe5c50d6c72865bce6657ea36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/f894d59db0ce07e81d6f780a4090821f45b5b3d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/f894d59db0ce07e81d6f780a4090821f45b5b3d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
