)]}'
{"id":"openstack%2Ffreezer-web-ui~694656","triplet_id":"openstack%2Ffreezer-web-ui~master~Iee12d4c186a373c5d8987dd48ece2b7a6eef446c","project":"openstack/freezer-web-ui","branch":"master","hashtags":[],"change_id":"Iee12d4c186a373c5d8987dd48ece2b7a6eef446c","subject":"Update devstack doc","status":"MERGED","created":"2019-11-16 01:29:07.000000000","updated":"2019-11-16 03:40:41.000000000","submitted":"2019-11-16 03:40:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694656-1573875641031-ee4c743d","meta_rev_id":"5a7d8f502032469f366d11df34fda6b9616b082c","_number":694656,"virtual_id_number":694656,"owner":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-16 03:40:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-16 03:36:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-16 03:36:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-16 03:40:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b17e1b8055c3c0ca6cff33bf3e6e32bcbd17c959","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2019-11-16 01:29:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a303eb9e4b3e4c93b5def5beb6636f8d053cd09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 02:09:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/300a1499d25c456aac730ae5f6d90e72 : SUCCESS in 5m 46s\n- freezer-web-ui-ubuntu https://zuul.opendev.org/t/openstack/build/dab293d10c9f4a86ace126835c9dff29 : SUCCESS in 39m 06s","accounts_in_message":[],"_revision_number":1},{"id":"d223f53f46be4794136091640186267dad910cdb","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2019-11-16 03:36:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3df86247feaba24011c880a6c744dade8a403c96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 03:37:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aeb0cf36d3a81fb3abcfdff0fc19cfd7ccd982d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 03:40:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b9a379cd5054d2c88926a3e673192e7 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"5a7d8f502032469f366d11df34fda6b9616b082c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 03:40:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1fc38359338b514fba631b32ceb6059b10f10485","revisions":{"1fc38359338b514fba631b32ceb6059b10f10485":{"kind":"REWORK","_number":1,"created":"2019-11-16 01:29:07.000000000","uploader":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"ref":"refs/changes/56/694656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/56/694656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/56/694656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/56/694656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/56/694656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/56/694656/1"}}},"commit":{"parents":[{"commit":"f894d59db0ce07e81d6f780a4090821f45b5b3d8","subject":"Add  py37 and remove py35 in envlst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/f894d59db0ce07e81d6f780a4090821f45b5b3d8"}]}],"author":{"name":"caihui","email":"cai.hui@zte.com.cn","date":"2019-11-16 01:26:49.000000000","tz":-480},"committer":{"name":"caihui","email":"cai.hui@zte.com.cn","date":"2019-11-16 01:26:49.000000000","tz":-480},"subject":"Update devstack doc","message":"Update devstack doc\n\nAdd a sampe \"local.conf\" file for freezer developer.\n\nChange-Id: Iee12d4c186a373c5d8987dd48ece2b7a6eef446c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/1fc38359338b514fba631b32ceb6059b10f10485"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/1fc38359338b514fba631b32ceb6059b10f10485"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
