)]}'
{"id":"openstack%2Ffreezer-web-ui~694919","triplet_id":"openstack%2Ffreezer-web-ui~master~I5c2ed1877aaec374efdbf0258ae9ec7a7e2d92dc","project":"openstack/freezer-web-ui","branch":"master","hashtags":[],"change_id":"I5c2ed1877aaec374efdbf0258ae9ec7a7e2d92dc","subject":"Replace git.openstack.org with opendev.org","status":"ABANDONED","created":"2019-11-19 07:36:27.000000000","updated":"2019-12-05 06:24:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e93b0f63845bb9e5cdccb179d85f00c5a3fe163c","_number":694919,"virtual_id_number":694919,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-19 08:06:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-19 08:06:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"80edcdf16d739af3062b7e1bbfff594fcd6978fc","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2019-11-19 07:36:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d77d15dbcab0df7f3b421b1462d9fbf8cb2be196","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 08:06:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/66e6b36958594d00bad0af02aebe1a52 : SUCCESS in 5m 08s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/f745da527f0b44779a2560ef386f0893 : SUCCESS in 5m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/daa0aaf3c1014ca4a60cd62f00818753 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91e69d1212054cb587272d54dbd92f32 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a8674441f465430193cdb4702b7f51dd : SUCCESS in 3m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4ae16aad87fd46ddb7a72241635f7152 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fa47202f68341ca9ce766f7e8bb440b : SUCCESS in 4m 52s\n- freezer-web-ui-ubuntu https://zuul.opendev.org/t/openstack/build/5ccf9bdefe114520ba60c1f8879a2d06 : FAILURE in 21m 30s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/26b7ba1b117841f08dc68e41dedafbb9 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e93b0f63845bb9e5cdccb179d85f00c5a3fe163c","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2019-12-05 06:24:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5502f76242f830576aa039596ecb6b2c91956847","revisions":{"5502f76242f830576aa039596ecb6b2c91956847":{"kind":"REWORK","_number":1,"created":"2019-11-19 07:36:27.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/19/694919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/19/694919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/19/694919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/19/694919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/19/694919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/19/694919/1"}}},"commit":{"parents":[{"commit":"1fc38359338b514fba631b32ceb6059b10f10485","subject":"Update devstack doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/1fc38359338b514fba631b32ceb6059b10f10485"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2019-11-19 07:24:08.000000000","tz":480},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2019-11-19 07:36:08.000000000","tz":480},"subject":"Replace git.openstack.org with opendev.org","message":"Replace git.openstack.org with opendev.org\n\nChange-Id: I5c2ed1877aaec374efdbf0258ae9ec7a7e2d92dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/5502f76242f830576aa039596ecb6b2c91956847"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/5502f76242f830576aa039596ecb6b2c91956847"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
