)]}'
{"id":"openstack%2Ffreezer-web-ui~700843","triplet_id":"openstack%2Ffreezer-web-ui~master~I499c5f2fb06caa70c939b6e503f65c14c2c2e474","project":"openstack/freezer-web-ui","branch":"master","topic":"drop-django111-support","hashtags":[],"change_id":"I499c5f2fb06caa70c939b6e503f65c14c2c2e474","subject":"Drop Django 1.11 support","status":"MERGED","created":"2020-01-01 20:21:10.000000000","updated":"2020-01-06 02:44:36.000000000","submitted":"2020-01-06 02:44:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700843-1578278676040-7a9322a4","meta_rev_id":"e24f89461d94eee4490286dd7428a7ab7819e237","_number":700843,"virtual_id_number":700843,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-01-06 02:01:20.000000000","_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-06 02:44:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-02 10:52:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-02 10:52:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-02 10:52:48.000000000","updated_by":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"reviewer":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"state":"REVIEWER"},{"updated":"2020-01-06 02:44:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc009ae1535a1a63f8590914323bdf62d8694682","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-01 20:21:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"290f1a2aa29ecea85b8a812fece23ed1e86bb901","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-01 20:50:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/85579908b599453b82e2450a5803d85b : SUCCESS in 3m 45s\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/57a93af9d81f4d2e8f590f7d62dcb1d9 : SUCCESS in 5m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39e06a0756c64cd1888d5ff3f61d2e3e : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9543c72235f4a9e9a062285b29fdf1a : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04f44b4d18794b91bf18b4dda6967574 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3367039cc2d44b7da34248fd90631330 : SUCCESS in 3m 57s\n- freezer-web-ui-ubuntu https://zuul.opendev.org/t/openstack/build/351ee5adbed94c629f7de75a72a5d350 : SUCCESS in 27m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/554b8d6981b1418087352bbfcc081f7a : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"bda4a1666acfefdfdd20ffc12a130056d53eb872","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-01-02 10:52:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM.\nThanks.","accounts_in_message":[],"_revision_number":1},{"id":"799ad26931286796ec0c91bdb0c6fb86177abf69","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-02 19:28:41.000000000","message":"Topic set to drop-django111-support","accounts_in_message":[],"_revision_number":1},{"id":"ce1de43eb2aaf0fb95d8c0329d8b1c978e9690a7","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-06 02:01:20.000000000","message":"Patch Set 1:\n\nrecheck.","accounts_in_message":[],"_revision_number":1},{"id":"2aa8d8d1723d7512100d97c26210952780a183f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 02:34:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/87027411767a44f69d32696cd4993238 : SUCCESS in 4m 03s\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/f9b0bbeba5214b25b1d31d66fb4948b5 : SUCCESS in 5m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d639149f9cfa41ed853135ac9457bb80 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9705b62c873243da8766b9fc33f1f57c : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f5a23e5d4bd4f709ba6c648b53f13d3 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ed91cfcfc0c44b3b5eac27482f00d69 : SUCCESS in 5m 58s\n- freezer-web-ui-ubuntu https://zuul.opendev.org/t/openstack/build/3dc7f3b5a4614bf88caa3fc703affa75 : SUCCESS in 32m 14s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/96d5e50744204d0495ca6cf36d60cecf : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"3693e2d665298e8d597ce41e65fec37fa761b2d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 02:35:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ab0f43d31a3ea23a8f3d9475f79a80c59a5ae9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 02:44:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/be7d49a6ce574636a598e10ca956a514 : SUCCESS in 2m 55s\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/c3c8aa02ee7443f382158812c2ac0469 : SUCCESS in 4m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a6de48865be4ef4af02eacf4f1e7e3a : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e242e90b2f24403b877f639e6d83796d : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/383df9783f4d45dbb4326f29b2c9a29a : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12befcc68c584b1fabcd13aaebcafd42 : SUCCESS in 3m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c8d7bd4d73cf47e197826cabb287841b : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e24f89461d94eee4490286dd7428a7ab7819e237","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 02:44:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"833a40b7b7de20408778a8175472ebcc2f807741","revisions":{"833a40b7b7de20408778a8175472ebcc2f807741":{"kind":"REWORK","_number":1,"created":"2020-01-01 20:21:10.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/43/700843/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/43/700843/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/43/700843/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/43/700843/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/43/700843/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/43/700843/1"}}},"commit":{"parents":[{"commit":"d0e560fc91028e6ab5c9802f30648bb8a2d1d80c","subject":"translation: drop babel extractor definitions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/d0e560fc91028e6ab5c9802f30648bb8a2d1d80c"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-01 20:20:33.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-01 20:21:08.000000000","tz":0},"subject":"Drop Django 1.11 support","message":"Drop Django 1.11 support\n\nDjango 1.11 ends its extended support in April 2020 (which is before\nUssuri release), so horizon drops Django 1.11 support in Ussuri.\n\ntox envs for non-primary Django versions are no longer needed in tox.ini\nas testing environments for non-primary Django versions are setup in\nthe zuul jobs now.\n\nhorizon\u003e\u003d17.1.0 is required to use Django 2.2. requirements.txt and\nlower-constraints.txt are updated accordingly. for more info. please\nrefer [1].\nDepends-On: https://review.opendev.org/#/c/700733/\n[1] https://etherpad.openstack.org/p/Enable_Django22_support_in_Horizon_Plugin\n\nChange-Id: I499c5f2fb06caa70c939b6e503f65c14c2c2e474\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/833a40b7b7de20408778a8175472ebcc2f807741"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/833a40b7b7de20408778a8175472ebcc2f807741"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
