)]}'
{"id":"openstack%2Ffreezer-web-ui~745360","triplet_id":"openstack%2Ffreezer-web-ui~master~I3d9cc639f5d5f523a1cdf395acacbfd88920e339","project":"openstack/freezer-web-ui","branch":"master","topic":"native-zuulv3-migration","hashtags":[],"change_id":"I3d9cc639f5d5f523a1cdf395acacbfd88920e339","subject":"zuul: migrate the freezer-web-ui job to native Zuul v3","status":"MERGED","created":"2020-08-07 15:29:15.000000000","updated":"2020-08-08 06:45:38.000000000","submitted":"2020-08-08 06:45:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745360-1596869138615-60d190a1","meta_rev_id":"d7890bfb25b70f79d57e1e7db10436f65bfc3f80","_number":745360,"virtual_id_number":745360,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-08 06:45:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-08 06:32:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-08 06:32:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-08 06:32:15.000000000","updated_by":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"reviewer":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"state":"REVIEWER"},{"updated":"2020-08-08 06:45:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"23715ded58a76da50fb675796178e0b10df3d4d7","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-07 15:29:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec0c4b2086d21ca4ab56e77abc9ad6deb8798bc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 15:29:30.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/freezer-web-ui on branch master.  The error was:\n\n  while parsing a block collection\n  in \"openstack/freezer-web-ui/.zuul.yaml@master\", line 21, column 7\ndid not find expected \u0027-\u0027 indicator\n  in \"openstack/freezer-web-ui/.zuul.yaml@master\", line 26, column 7","accounts_in_message":[],"_revision_number":1},{"id":"5cfbe6c1ccde207e9e7b007bc33fc63dd5529169","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-07 15:41:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7dc13d8c02f48f13ee7bf8efb76e95c01a709745","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 16:14:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74c87d14e6f5445494851d97a4c50467 : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14e8f44cab9343ad987742cef94c9efe : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a9dca81672047b6b93804f2eecf5492 : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca68615d795444cf880c9932cb7b1e38 : SUCCESS in 7m 26s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/404888c8b7254cf2996d1c649d630e41 : SUCCESS in 30m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/10b003d4480d4e2e8bc67da6716c9fc5 : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":2},{"id":"1ad61db807bdefec9549ce38b623655d91ef91ca","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-07 16:18:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f7ef0f0c8b8b3a2edb51c1b5366e6b9b5d1b00df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 17:01:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49c09f0edc9942e0a79e6def43db363f : SUCCESS in 7m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0d3f56ba2d343c5bf48dfb9f16a168e : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c9a470905174e62bc8d4b6b371c84cb : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c28cb51ae577438a8c8d9dbfd101bb1d : SUCCESS in 6m 40s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/605b0ce072d14fa29cd24134baca3192 : SUCCESS in 35m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0aee04fc3736425eab208628d2a49894 : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":3},{"id":"018003a628c81cd66e498584f81b73ae5f24c262","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-08-08 06:31:59.000000000","message":"Patch Set 3:\n\nLGTM. Thanks.","accounts_in_message":[],"_revision_number":3},{"id":"ae8cc5b9dc6458879472c52b91f14c975475414e","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-08-08 06:32:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLGTM. Thanks.","accounts_in_message":[],"_revision_number":3},{"id":"a9094b0128bb800d5d90f0d581141a208c87244c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 06:32:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c5faeda6759b7f32ea168c61d1d1d40228ca9695","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 06:45:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edcaf189aebf4bb8ba61be3b8c9baff5 : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01e9fc8d2c1140609cb64ae115d86779 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4fe27fbba5440d489d68266bfbacd29 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2a6a005c04d480cb2c85ec3e84cfb6b : SUCCESS in 6m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/66e8e5127c70466a89f9bfbba77cf5e8 : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":3},{"id":"767765afb520726f0bf385e41c36fb1f7fc76e2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 06:45:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1f5004d5214e46064ba6f3444ff813715e7808ff","revisions":{"273fbe050da110846b18328dc24a7ad2c7318e99":{"kind":"REWORK","_number":1,"created":"2020-08-07 15:29:15.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/60/745360/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/60/745360/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/60/745360/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/60/745360/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/60/745360/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/60/745360/1"}}},"commit":{"parents":[{"commit":"e57a654832736adb750e9a2e340a51fd21b2857f","subject":"Add py38 in tox.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/e57a654832736adb750e9a2e340a51fd21b2857f"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-07 15:28:23.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-07 15:28:23.000000000","tz":120},"subject":"zuul: migrate the freezer-web-ui job to native Zuul v3","message":"zuul: migrate the freezer-web-ui job to native Zuul v3\n\nThe current job only checks whether the repository can be installed.\nIt specifies a tempest regex, but it is invalid, and anyway\nthe tempest tests do not touch horizon and its plugins, so\nthe job can just inherit from devstack instead of devstack-tempest.\n\nDrop the distribution-specific suffix, ubuntu is the default.\n\nChange-Id: I3d9cc639f5d5f523a1cdf395acacbfd88920e339\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/273fbe050da110846b18328dc24a7ad2c7318e99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/273fbe050da110846b18328dc24a7ad2c7318e99"}]},"branch":"refs/heads/master"},"6d694bceefbd7731ea396ad9aa4aa49aea983530":{"kind":"REWORK","_number":2,"created":"2020-08-07 15:41:12.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/60/745360/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/60/745360/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/60/745360/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/60/745360/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/60/745360/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/60/745360/2"}}},"commit":{"parents":[{"commit":"e57a654832736adb750e9a2e340a51fd21b2857f","subject":"Add py38 in tox.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/e57a654832736adb750e9a2e340a51fd21b2857f"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-07 15:41:03.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-07 15:41:03.000000000","tz":120},"subject":"zuul: migrate the freezer-web-ui job to native Zuul v3","message":"zuul: migrate the freezer-web-ui job to native Zuul v3\n\nThe current job only checks whether the repository can be installed.\nIt specifies a tempest regex, but it is invalid, and anyway\nthe tempest tests do not touch horizon and its plugins, so\nthe job can just inherit from devstack instead of devstack-tempest.\n\nDrop the distribution-specific suffix, ubuntu is the default.\n\nChange-Id: I3d9cc639f5d5f523a1cdf395acacbfd88920e339\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/6d694bceefbd7731ea396ad9aa4aa49aea983530"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/6d694bceefbd7731ea396ad9aa4aa49aea983530"}]},"branch":"refs/heads/master"},"1f5004d5214e46064ba6f3444ff813715e7808ff":{"kind":"REWORK","_number":3,"created":"2020-08-07 16:18:59.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/60/745360/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/60/745360/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/60/745360/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/60/745360/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/60/745360/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/60/745360/3"}}},"commit":{"parents":[{"commit":"e57a654832736adb750e9a2e340a51fd21b2857f","subject":"Add py38 in tox.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/e57a654832736adb750e9a2e340a51fd21b2857f"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-07 16:18:49.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-07 16:18:49.000000000","tz":120},"subject":"zuul: migrate the freezer-web-ui job to native Zuul v3","message":"zuul: migrate the freezer-web-ui job to native Zuul v3\n\nThe current job only checks whether the repository can be installed.\nIt specifies a tempest regex, but it is invalid, and anyway\nthe tempest tests do not touch horizon and its plugins, so\nthe job can just inherit from devstack instead of devstack-tempest.\n\nDrop the distribution-specific suffix, ubuntu is the default.\n\nChange-Id: I3d9cc639f5d5f523a1cdf395acacbfd88920e339\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/1f5004d5214e46064ba6f3444ff813715e7808ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/1f5004d5214e46064ba6f3444ff813715e7808ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
