)]}'
{"id":"openstack%2Ffreezer-web-ui~747165","triplet_id":"openstack%2Ffreezer-web-ui~master~Iab13d5fb6fd4c6239ab34281e16830545c41bef0","project":"openstack/freezer-web-ui","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"Iab13d5fb6fd4c6239ab34281e16830545c41bef0","subject":"[DNM] Test freezer-web-ui with horizon jobs on focal","status":"ABANDONED","created":"2020-08-20 13:46:20.000000000","updated":"2020-09-24 07:07:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"222e5440e333d6891b5fddd32ddf0d0ee9925c89","_number":747165,"virtual_id_number":747165,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-09-23 18:04:47.000000000","_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-23 18:55:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-11 11:15:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 11:15:33.000000000","updated_by":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"reviewer":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"state":"REVIEWER"},{"updated":"2020-09-23 18:55:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"45485608ceb5f7214b04c0a095f41792a21ed361","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-08-20 13:46:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07424de9674f3435c248eb07cf8324dff4d152cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 14:26:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c60dcd8bdeaf4cbd882b427cc1eb28af : SUCCESS in 5m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/143f7b7015ea42f7a0759bdb9aa3a9d9 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df9535e6913047f7893bd2ff3e04f7b7 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18539982e48b4221b980de58e58cc942 : RETRY_LIMIT in 2m 31s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/c2887afc6f4a4c1491afb2ce3721f65e : FAILURE in 32m 43s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fd75b595c57c42daa3d6e41f1809fc68 : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ad979cae7e26c474453483ebc3e44d432cabc6a8","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-08-20 14:34:08.000000000","message":"Topic 687799 removed","accounts_in_message":[],"_revision_number":1},{"id":"5192fbacfd567f2c353ed543d4a54ee7f3526f4f","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-08-20 14:35:30.000000000","message":"Patch Set 1:\n\nrecheck.","accounts_in_message":[],"_revision_number":1},{"id":"5f02b0182e63ec34c82a1a5b0ae67eaf825db5f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 16:34:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fab0d21f08643699a80a0ebbb4cdfcd : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c45bbdde474b4ecf99a041e554129ca7 : RETRY_LIMIT in 1m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cc7b0c745454bac8e98650489ab7e70 : RETRY_LIMIT in 2m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad9d26e42125437f85fe02315b360634 : RETRY_LIMIT in 2m 15s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/d0be66d94a594621aa4a6a2a367c2092 : RETRY_LIMIT in 2m 07s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e5f1490194374923acbac7ca606032a0 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"33e527b3e0cd5fd812741803ff0c73dca0fc1beb","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-08-20 17:02:01.000000000","message":"Patch Set 1:\n\nrecheck.","accounts_in_message":[],"_revision_number":1},{"id":"5dcb94d10948b113a19c3bdb826009ed24d091a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 18:03:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aac4503c0b534ba2a8dc22cbc6a79cb9 : SUCCESS in 6m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1500e01755b4601bfac2c608c6bb9b1 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/995d3ff45c93406d9035448081679730 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c875e170c3fc41969503830d87f6e07e : SUCCESS in 6m 39s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/0f40e7b9c635433bbd869cd17bdabfb1 : FAILURE in 35m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/311b86047dbe4afca0ec8019a6c06994 : SUCCESS in 8m 08s","accounts_in_message":[],"_revision_number":1},{"id":"672e36cbcd30a0ccaf199301abf22092316b49fc","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-08-21 11:16:28.000000000","message":"Patch Set 1:\n\nrecheck.","accounts_in_message":[],"_revision_number":1},{"id":"e89550d7c6571e7454a5a756439f9c84aa16f3d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 12:03:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01c417da6c924ccfb29cffe82a0faf98 : SUCCESS in 3m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60c8867d4b074c3e9f46ff402c24cfe3 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e21fc70933a49adbd776888cf9eff83 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26bde650a54b458e9bf53c09c32b675e : SUCCESS in 4m 55s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/9e073238005247dda023f81b6c9d1ebe : FAILURE in 35m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e7cb9e9ec98344479ebb9878ce3c8e87 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":1},{"id":"8101cd629d856617e16bdf6f478bb3aed0819749","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-08-21 12:06:53.000000000","message":"Patch Set 1:\n\nrecheck.","accounts_in_message":[],"_revision_number":1},{"id":"ef4dd26eeb7a6f01a03bd42273c8446448dfffbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 12:53:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ff98b7b5f704b6d8921623bf903d3e7 : SUCCESS in 4m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ef649b4ca45432eba2a0df724760908 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48b959a6c994427cb2cc311293a46f81 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c07d063189074c108edf5a7dad5edc4f : SUCCESS in 3m 56s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/6f00e2e8ce6c470291abafc4f95f196f : FAILURE in 37m 27s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/71dc5ddefc4c450ebc720ac721f77678 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f213bd0309bf774897b31d23b0d4248db9ed5cf2","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-23 18:04:40.000000000","message":"Topic set to migrate-to-focal","accounts_in_message":[],"_revision_number":1},{"id":"9d10e24e1219e417ffe013ba8060ae5a59288c54","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-23 18:04:47.000000000","message":"Patch Set 1:\n\nrecheck.","accounts_in_message":[],"_revision_number":1},{"id":"af4d5712273e7e621ef49665a722d3df8d96525d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 18:55:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/133ba4054d374408b9e7c4d0f0290fcc : SUCCESS in 7m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/804e9720f0454c4fb7d1ea87a86e6d6f : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0214d66fa545481fb96b66f0f5a42049 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfbd916038dc46ff901adf32bfc34ac4 : SUCCESS in 4m 57s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/da8352871b2d43c69d4c3f940e4297eb : SUCCESS in 41m 43s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/300b3eb89d4340c793a3a00b73b19f93 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"222e5440e333d6891b5fddd32ddf0d0ee9925c89","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-24 07:07:22.000000000","message":"Abandoned\n\nAll looks good here.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5479d973fb35120ab8ba95898ac1e0c340a39ac9","revisions":{"5479d973fb35120ab8ba95898ac1e0c340a39ac9":{"kind":"REWORK","_number":1,"created":"2020-08-20 13:46:20.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/65/747165/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/65/747165/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/65/747165/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/65/747165/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/65/747165/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/65/747165/1"}}},"commit":{"parents":[{"commit":"75b145a459a544b4a0b769e4364703da4a87f856","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/75b145a459a544b4a0b769e4364703da4a87f856"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-08-20 13:45:01.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-08-20 13:45:01.000000000","tz":0},"subject":"[DNM] Test freezer-web-ui with horizon jobs on focal","message":"[DNM] Test freezer-web-ui with horizon jobs on focal\n\nDepends-On: https://review.opendev.org/#/c/744647/\nChange-Id: Iab13d5fb6fd4c6239ab34281e16830545c41bef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/5479d973fb35120ab8ba95898ac1e0c340a39ac9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/5479d973fb35120ab8ba95898ac1e0c340a39ac9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
