)]}'
{"id":"openstack%2Ffreezer-web-ui~778362","triplet_id":"openstack%2Ffreezer-web-ui~master~Ie2567b7f54e8cc54e1814b6884a5184581e1bd5a","project":"openstack/freezer-web-ui","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"Ie2567b7f54e8cc54e1814b6884a5184581e1bd5a","subject":"Imported Translations from Zanata","status":"MERGED","created":"2021-03-03 06:27:21.000000000","updated":"2021-03-15 06:58:44.000000000","submitted":"2021-03-15 06:58:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778362-zanata/translations","meta_rev_id":"55ef2ebebd074a4c8b95c134548866d56315560e","_number":778362,"virtual_id_number":778362,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-15 06:58:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"recommended":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":1,"date":"2021-03-03 18:49:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2021-03-04 20:45:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-11 00:48:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-15 06:32:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-03 06:43:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 18:49:33.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2021-03-04 20:45:56.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2021-03-10 15:28:42.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"state":"REVIEWER"}],"messages":[{"id":"e5a95fa4dab3808bbe8a61d77936b49a637a2fc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-03 06:27:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb9883ca8d60038476b4eca3784c515a0f77d42e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 06:43:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfb51c5440a34462873b4d85e208b3ba : SUCCESS in 7m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c140a0a9fef34b848b7fa0d2eb0120ed : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f34f6ab851f4ad6b96f7b92e25c8b0b : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/909193b5474d4b6482b60dcc58b9a6b1 : SUCCESS in 7m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75db10d0dfc14167801cb30d7d7e64c8 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e02769fdf0cb4cf5990adbab0b4b0441 : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":1},{"id":"37961d77893ccae5292cf40bb8b3765ee8f1c2b9","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-03-03 18:49:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9672115722de4f0f9d1171f3414f24d6b3522aec","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2021-03-04 20:45:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e25f274e9813dd77dd66fe5f9e5ad3756f888b82","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2021-03-11 00:48:17.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM.\nThanks.","accounts_in_message":[],"_revision_number":1},{"id":"9122ad95798e0eb3f9327ae667d6af7394fbb0d6","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2021-03-15 06:32:51.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e95b9964d772499eaa25f1e3ff040a87c681156a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 06:33:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"616e62efc85cc4f9bcf035ca3762692952f67423","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 06:58:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0eeeaaf6f0584217bb9f4fb17314a9f7 : SUCCESS in 6m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ad66321fc6d41589c0902b302f8b6a3 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c5ecd86db4e483784a26411a1659550 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1af63ba7919e46eebecd36b9ea0bf66b : SUCCESS in 6m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/636ef34b566f4e4a89b72d5b2f0d48b3 : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":1},{"id":"55ef2ebebd074a4c8b95c134548866d56315560e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 06:58:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1c0d57ede8f49dddf5a625db27a7ac387c3f0cb","revisions":{"c1c0d57ede8f49dddf5a625db27a7ac387c3f0cb":{"kind":"REWORK","_number":1,"created":"2021-03-03 06:27:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/62/778362/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/62/778362/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/62/778362/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/62/778362/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/62/778362/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/62/778362/1"}}},"commit":{"parents":[{"commit":"b4f7b7ae78e3d1567ab1eec0f1532f2b1143482c","subject":"Use TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/b4f7b7ae78e3d1567ab1eec0f1532f2b1143482c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-03 06:27:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-03 06:27:19.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ie2567b7f54e8cc54e1814b6884a5184581e1bd5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/c1c0d57ede8f49dddf5a625db27a7ac387c3f0cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/c1c0d57ede8f49dddf5a625db27a7ac387c3f0cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
