)]}'
{"id":"openstack%2Ffreezer-web-ui~973872","triplet_id":"openstack%2Ffreezer-web-ui~master~Ia0cf6af38d9e0820e03a216286c193ddd0f9d235","project":"openstack/freezer-web-ui","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-01-20 08:52:50.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-20 09:22:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia0cf6af38d9e0820e03a216286c193ddd0f9d235","subject":"Fix devstack horizon plugin enablement","status":"MERGED","created":"2026-01-19 17:07:31.000000000","updated":"2026-02-03 18:22:49.000000000","submitted":"2026-01-20 09:22:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973872","meta_rev_id":"159639ffbdf1d242b6f2cd8fd7242fd984927730","_number":973872,"virtual_id_number":973872,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-20 09:22:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-01-20 08:52:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-01-20 08:52:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-19 17:47:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-19 19:43:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"ee1082276653603346d0703fe932cbf2707877ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-19 17:07:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6d07a63454aeac2bfd2ad5ff1cc081cf6617047","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 17:47:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71b635234b3c42aabc219c950ce1b941\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5435b558f08449d69506bef4ae2fe533 : SUCCESS in 5m 14s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/5e73f836a95a471d904f81e2c7e47cb0 : SUCCESS in 5m 28s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/6dd1c6ef22204a49a4d8e27ac1ac1fc7 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbe4bceeaeff41d0b206605f09ac4a04 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a62d9ad67484ac7a6b7d2318976f6bd : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a2867630d384aa99b769425f64d324f : SUCCESS in 8m 30s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/c227122f71164605924d434d767c7504 : SUCCESS in 34m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/807031b7a33f4c0093b15345481bf567 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"089fad6f6a15a3a3fd7d79d549ae339cf9f93875","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 19:43:25.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a79709fc478281d5a6c2ee48c146f2a00215e19","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 19:48:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"45d181aec6e706a09cfee0b6224c97c5762f6186","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-20 08:45:01.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"fc0f152c572ff4737c92f0cbed329507a558c402","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-20 08:52:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c61ac3b794cf2c4354d001e6b679158705a1f2f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 09:17:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98c3ecc72b3447a19c4d9ad07979fed9\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8fe2c2f8a06a40ca8b41d3999a19be10 : SUCCESS in 2m 21s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/ea36a5d6027e47dc867f7d54cf1b8c66 : SUCCESS in 6m 04s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/cfbf698f2dcc45d7a05aed446814fafb : SUCCESS in 2m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da6ddbd2ebfb4375ba144016aafec0a3 : SUCCESS in 2m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83a188dc6f1d4a4a89da9f1d6832ff97 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2edebc10d54f4c4bb9ec0664b3223561 : SUCCESS in 7m 54s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/f5ef288120974975adc890f707be6fb1 : SUCCESS in 27m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8d3c25be497e45ca9cb58188b6dfb345 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":2},{"id":"0a3304c6ae9fcc0109edf16c703b787ad0f763d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 09:17:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bd97a0858162906e4d63370ea59bd2bca6130836","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 09:22:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc3fa56d6e6743bc847d7787cd204b71\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/890f0cf04223461fb061733cb8f2fbdc : SUCCESS in 2m 17s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/6f72e4ffd64046c984425c09b01a3558 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab888bb152e34409ab78b061c5136bf7 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7868d0562f745ba9368b8c8ec06da53 : SUCCESS in 2m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97870b02218a4fe082085765eedff257 : SUCCESS in 3m 50s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/678841bec2a74d3c96f79c533cbc12fe : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":2},{"id":"e91ced3dba005b4026bf71e9bc4167e8ea729df9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 09:22:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ce3e6fc7c1d7d349e55dde197ebca720033950c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 09:23:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed9a03f0a67a46e8838ec931e10d2319\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d0c7fea7ce843e5869a06b6fca09a26 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2},{"id":"159639ffbdf1d242b6f2cd8fd7242fd984927730","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-03 18:22:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"835b3ed93c4919ca3f975a34744c7e92b8e93297","revisions":{"32853d38b13f6363ba883ec447e7367830a66e69":{"kind":"REWORK","_number":1,"created":"2026-01-19 17:07:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/973872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/72/973872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973872/1"}}},"commit":{"parents":[{"commit":"b69d1a61ea38e1a8951198bda8a662fb3f099b98","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/b69d1a61ea38e1a8951198bda8a662fb3f099b98"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-01-19 17:07:25.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-01-19 17:07:25.000000000","tz":60},"subject":"Fix devstack horizon pllugin enablement","message":"Fix devstack horizon pllugin enablement\n\nAfter renaming of the module enabled file was fixed, but\nnot the one for devstack, which causes horizon to fail with 500 on\nimport attempt.\n\nChange-Id: Ia0cf6af38d9e0820e03a216286c193ddd0f9d235\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/32853d38b13f6363ba883ec447e7367830a66e69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/32853d38b13f6363ba883ec447e7367830a66e69"}]},"branch":"refs/heads/master"},"835b3ed93c4919ca3f975a34744c7e92b8e93297":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-01-20 08:45:01.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/973872/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/72/973872/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973872/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973872/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973872/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973872/2"}}},"commit":{"parents":[{"commit":"b69d1a61ea38e1a8951198bda8a662fb3f099b98","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/b69d1a61ea38e1a8951198bda8a662fb3f099b98"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-01-19 17:07:25.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-01-20 08:45:01.000000000","tz":0},"subject":"Fix devstack horizon plugin enablement","message":"Fix devstack horizon plugin enablement\n\nAfter renaming of the module enabled file was fixed, but\nnot the one for devstack, which causes horizon to fail with 500 on\nimport attempt.\n\nChange-Id: Ia0cf6af38d9e0820e03a216286c193ddd0f9d235\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/835b3ed93c4919ca3f975a34744c7e92b8e93297"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/835b3ed93c4919ca3f975a34744c7e92b8e93297"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
