)]}'
{"id":"openstack%2Ffreezer-web-ui~973972","triplet_id":"openstack%2Ffreezer-web-ui~master~Ic7a274a90715a5a745a229d8b57ea7e4968921c5","project":"openstack/freezer-web-ui","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-02-03 17:48:55.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Ic7a274a90715a5a745a229d8b57ea7e4968921c5","subject":"Register a panel for API url","status":"MERGED","created":"2026-01-20 17:36:06.000000000","updated":"2026-02-03 18:19:52.000000000","submitted":"2026-02-03 18:19:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973972","meta_rev_id":"c656e03b8b8c8ac3359943d38bb7763443dbf9eb","_number":973972,"virtual_id_number":973972,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-03 18:19:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-02-03 17:48:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-02 16:00:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-02-03 17:48:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-20 18:14:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-20 20:01:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"4d4c1c69255e302b5a9bb681294eb281b7addf0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-20 17:36:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"edee81481e3504ec60f24d205a52a2884a242e52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-20 17:37:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f018c3c15755b21e82739f3635714bf6306c9693","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-20 17:44:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e90ac4cbd5ee37c5244184ccd792fc161b18b11a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-20 17:45:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e53a8deafee245eca93adea175ae79888abf6d91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 18:14:41.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0dd2c867fcd94424a3cd3437888c4273\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c87db3ea4b443a58d544608a8b0053a : SUCCESS in 3m 55s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/a0538bfd3d5840079befae2637d905c6 : SUCCESS in 4m 21s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/2be808724d1544beb02e34b7ea3dd09c : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ae831595b74420e805739e9f5bb5e0a : FAILURE in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0fcfd578aacd43c18c88b12c1cc66d38 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7fec952e3774f779a75192a6c4f9b63 : SUCCESS in 7m 29s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/3f8d90e47aff49a3b5120d14985fb1ba : SUCCESS in 25m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1f0e9af360e74404a08ef2919a303419 : FAILURE in 3m 28s","accounts_in_message":[],"_revision_number":4},{"id":"2c9dd465c987e7e7506624972381eaa33c8109bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-20 18:26:27.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"cf29ffcb9f93b494cac8727dd04d6b45c7de649c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 18:55:29.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f6df8f7f3a24b929384ba21264391b3\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ac62adc6c7144ac9b0e6d4e6f82eca0 : SUCCESS in 7m 03s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/e2d401e24b3b4ca68010be6a2f4a09f0 : SUCCESS in 4m 53s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/27db9d929c0345319bdbb3c6d85707b4 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6a171a7c4994312b720c2d2397c9b93 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7054969b6a34226bd551362c591f6fc : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/215e0027a73b4a89837c58c5d31abde1 : SUCCESS in 8m 42s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/d92648cbc41349b8b11c4cc25cdd926e : SUCCESS in 12m 15s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5095e75cd8e1428c84f32d5874c18966 : FAILURE in 2m 42s","accounts_in_message":[],"_revision_number":5},{"id":"6310550897cbb320590a1fbad6dd99fb34205699","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-20 19:27:55.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"f057e5c53ba635bc6f4205a020ea0bdc4ebce9ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 19:51:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd4907ec22ad48508496d6506bc5ca10\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aebc9346ea804d45bbe27c2047d00419 : SUCCESS in 2m 33s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/2d082d00d60c499195a923c1ecaf77d2 : SUCCESS in 4m 40s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/05e6ade791b14ec68378a69777a7ee8e : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d792677e7bfc430cbcb43a243a1f2fbc : SUCCESS in 2m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa73caa83ba44fc894c718d6ced11016 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6835b84010564dc78b49a9b87ff2f66e : SUCCESS in 4m 18s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/1212c51f23124cdd90639842188466b8 : SUCCESS in 13m 14s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/33c87198f00549c5a1d910e22c7ae077 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":6},{"id":"a0f62a9437ddd348464aa16cee02c090e5d049e4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-20 20:01:10.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4cac63e8e295164f6be4d03930944101b32e3323","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-02 16:00:48.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"9dec5dbd1e6b594bfd73314671fb2298dfa74e61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 16:29:10.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dd5a814c9da45049bb50aafe1643cd7\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb75c8a3caba4d3c83703d49a5e1a161 : SUCCESS in 7m 05s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/2acfed7d722f4bb194a21039887a2c00 : SUCCESS in 4m 46s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/f9dcd63670b94748adfc6d067425c9f6 : SUCCESS in 2m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e74f52a283ff4991a2b4c8bc9a1dad12 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e25b71995fb40b6b1d3868f98fe6ddb : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44a6ac64285746c6a006cc8330bc057b : SUCCESS in 7m 54s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/7bb051f6864d4e5ba347f89edaf78ac1 : SUCCESS in 14m 07s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5c581de69ffc457db8c82af0aea52bcf : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":7},{"id":"72b3d8f5afa05f14aa8ce69e66bc9c1e327327eb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-03 17:48:55.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"a198441f48fc2f86f43193e8aaefd8b8a6955af3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 17:49:49.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"83c9fc33408d88528c516147563a4e50e11dbf24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 18:19:52.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1a148b7ccf541c3ac160575be402596\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/059cefb5a5e14587aacb59f728aa6854 : SUCCESS in 2m 41s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/acb200cbac0b4cc7ba32f5b194a0b310 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c87259994584d359a4cada82d93913d : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c62d475935a40919b2117f0793b4e57 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04c7aa8d3d0844438f8e67a666b7e896 : SUCCESS in 5m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/27e0cb6863994308adb53227bff0ad79 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":7},{"id":"c656e03b8b8c8ac3359943d38bb7763443dbf9eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 18:19:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"f90df7eaf4a928b47b039d4079c73b61ff03eddc","revisions":{"0ff3a2f8016e95f923980aa7c2758e66dbb983c5":{"kind":"REWORK","_number":1,"created":"2026-01-20 17:36:06.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/973972/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/72/973972/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/1"}}},"commit":{"parents":[{"commit":"478a5a7c0d08e1c2f1094bc0cd6196dcdbfd4f82","subject":"Move freezer panel under Projects dashboard","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/478a5a7c0d08e1c2f1094bc0cd6196dcdbfd4f82"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-20 17:35:58.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-20 17:35:58.000000000","tz":60},"subject":"Register a panel for API url","message":"Register a panel for API url\n\nWith removal of the dashboard, URLs for API are not registered anymore.\n\nFor registering API URLs we need to add another panel, which\nwill not contain any content, and thus will not show up in Horizon,\nwhile being properly registered.\n\nChange-Id: Ic7a274a90715a5a745a229d8b57ea7e4968921c5\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/0ff3a2f8016e95f923980aa7c2758e66dbb983c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/0ff3a2f8016e95f923980aa7c2758e66dbb983c5"}]},"branch":"refs/heads/master"},"9a395bfed706b0dd55ed740e84647a3aeca2f991":{"kind":"REWORK","_number":2,"created":"2026-01-20 17:37:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/973972/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/72/973972/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/2"}}},"commit":{"parents":[{"commit":"478a5a7c0d08e1c2f1094bc0cd6196dcdbfd4f82","subject":"Move freezer panel under Projects dashboard","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/478a5a7c0d08e1c2f1094bc0cd6196dcdbfd4f82"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-20 17:35:58.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-20 17:36:58.000000000","tz":60},"subject":"Register a panel for API url","message":"Register a panel for API url\n\nWith removal of the dashboard, URLs for API are not registered anymore.\n\nFor registering API URLs we need to add another panel, which\nwill not contain any content, and thus will not show up in Horizon,\nwhile being properly registered.\n\nChange-Id: Ic7a274a90715a5a745a229d8b57ea7e4968921c5\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/9a395bfed706b0dd55ed740e84647a3aeca2f991"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/9a395bfed706b0dd55ed740e84647a3aeca2f991"}]},"branch":"refs/heads/master"},"76f68250947d57d16f62fc1143f543ddbfa6f2bc":{"kind":"REWORK","_number":3,"created":"2026-01-20 17:44:09.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/973972/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/72/973972/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/3"}}},"commit":{"parents":[{"commit":"4a990b0515a3873612d5be6cf80ea6f9cc4c0b8c","subject":"Move freezer panel under Projects dashboard","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/4a990b0515a3873612d5be6cf80ea6f9cc4c0b8c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-20 17:35:58.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-20 17:43:09.000000000","tz":60},"subject":"Register a panel for API url","message":"Register a panel for API url\n\nWith removal of the dashboard, URLs for API are not registered anymore.\n\nFor registering API URLs we need to add another panel, which\nwill not contain any content, and thus will not show up in Horizon,\nwhile being properly registered.\n\nChange-Id: Ic7a274a90715a5a745a229d8b57ea7e4968921c5\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/76f68250947d57d16f62fc1143f543ddbfa6f2bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/76f68250947d57d16f62fc1143f543ddbfa6f2bc"}]},"branch":"refs/heads/master"},"de715d9d8eb317e36efea564b193f9792815f65a":{"kind":"REWORK","_number":4,"created":"2026-01-20 17:45:09.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/973972/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/72/973972/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/4"}}},"commit":{"parents":[{"commit":"4a990b0515a3873612d5be6cf80ea6f9cc4c0b8c","subject":"Move freezer panel under Projects dashboard","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/4a990b0515a3873612d5be6cf80ea6f9cc4c0b8c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-20 17:35:58.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-20 17:45:01.000000000","tz":60},"subject":"Register a panel for API url","message":"Register a panel for API url\n\nWith removal of the dashboard, URLs for API are not registered anymore.\n\nFor registering API URLs we need to add another panel, which\nwill not contain any content, and thus will not show up in Horizon,\nwhile being properly registered.\n\nChange-Id: Ic7a274a90715a5a745a229d8b57ea7e4968921c5\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/de715d9d8eb317e36efea564b193f9792815f65a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/de715d9d8eb317e36efea564b193f9792815f65a"}]},"branch":"refs/heads/master"},"c091f79927804868715d1a903ada9634a77e1540":{"kind":"REWORK","_number":5,"created":"2026-01-20 18:26:27.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/973972/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/72/973972/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/5"}}},"commit":{"parents":[{"commit":"4a990b0515a3873612d5be6cf80ea6f9cc4c0b8c","subject":"Move freezer panel under Projects dashboard","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/4a990b0515a3873612d5be6cf80ea6f9cc4c0b8c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-20 17:35:58.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-20 18:26:27.000000000","tz":0},"subject":"Register a panel for API url","message":"Register a panel for API url\n\nWith removal of the dashboard, URLs for API are not registered anymore.\n\nFor registering API URLs we need to add another panel, which\nwill not contain any content, and thus will not show up in Horizon,\nwhile being properly registered.\n\nChange-Id: Ic7a274a90715a5a745a229d8b57ea7e4968921c5\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/c091f79927804868715d1a903ada9634a77e1540"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/c091f79927804868715d1a903ada9634a77e1540"}]},"branch":"refs/heads/master"},"1572a15bfb5f2208664ca6ce19f603240b824b76":{"kind":"REWORK","_number":6,"created":"2026-01-20 19:27:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/973972/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/72/973972/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/6"}}},"commit":{"parents":[{"commit":"4a990b0515a3873612d5be6cf80ea6f9cc4c0b8c","subject":"Move freezer panel under Projects dashboard","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/4a990b0515a3873612d5be6cf80ea6f9cc4c0b8c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-20 17:35:58.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-20 19:27:49.000000000","tz":60},"subject":"Register a panel for API url","message":"Register a panel for API url\n\nWith removal of the dashboard, URLs for API are not registered anymore.\n\nFor registering API URLs we need to add another panel, which\nwill not contain any content, and thus will not show up in Horizon,\nwhile being properly registered.\n\nChange-Id: Ic7a274a90715a5a745a229d8b57ea7e4968921c5\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/1572a15bfb5f2208664ca6ce19f603240b824b76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/1572a15bfb5f2208664ca6ce19f603240b824b76"}]},"branch":"refs/heads/master"},"f90df7eaf4a928b47b039d4079c73b61ff03eddc":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-02-02 16:00:48.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/973972/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/72/973972/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/72/973972/7"}}},"commit":{"parents":[{"commit":"5e8853aa29dc43e94c35f9aef3cf41b7e46b7177","subject":"Move freezer panel under Projects dashboard","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/5e8853aa29dc43e94c35f9aef3cf41b7e46b7177"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-20 17:35:58.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-02 16:00:48.000000000","tz":0},"subject":"Register a panel for API url","message":"Register a panel for API url\n\nWith removal of the dashboard, URLs for API are not registered anymore.\n\nFor registering API URLs we need to add another panel, which\nwill not contain any content, and thus will not show up in Horizon,\nwhile being properly registered.\n\nChange-Id: Ic7a274a90715a5a745a229d8b57ea7e4968921c5\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/f90df7eaf4a928b47b039d4079c73b61ff03eddc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/f90df7eaf4a928b47b039d4079c73b61ff03eddc"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
