)]}'
{"id":"openstack%2Ffreezer-web-ui~975893","triplet_id":"openstack%2Ffreezer-web-ui~master~I57156d0f3fdb0702431d14b35f58266b68853d8b","project":"openstack/freezer-web-ui","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-06-02 08:40:47.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-06-02 08:27:46.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I57156d0f3fdb0702431d14b35f58266b68853d8b","subject":"Add interval unit/value selector to session creation","status":"MERGED","created":"2026-02-06 11:19:34.000000000","updated":"2026-06-02 08:41:41.000000000","submitted":"2026-06-02 08:40:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975893","meta_rev_id":"9cee9c70754c5d12ee7947470b4b3cd859af75da","_number":975893,"virtual_id_number":975893,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 08:40:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-06-02 08:27:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-06-02 08:27:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-06 11:41:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-19 22:19:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-04-19 22:19:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"42a9f1d8b0d4a2697d2ef8b70ed7a4bea4c80dda","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-06 11:19:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6a12dd0f4187c32c768b9fe05b570e75e7a3b09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 11:41:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0094210a1697460cac03d4a8d86b03fc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8b6f9c2ca7cb42f3b2196bd2ef1475e0 : SUCCESS in 2m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/48121c584a3348829346c90512b6bfbd : SUCCESS in 2m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b2d6c2a9fff41b7964b32139d60ae55 : SUCCESS in 4m 26s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/616e9decf36340bbaa8d88d0ede7d0e3 : SUCCESS in 4m 38s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/fa457c85c4df4218be66f4a9a6e917fd : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01dd7f362d754e8b876611eb6bb1c2d7 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1f53f31607343fba6f94d3816835045 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e41956e8df5446a1a5dd73ad0194403f : SUCCESS in 7m 58s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/1597ef4bab084797bae28b96c73dbeac : SUCCESS in 19m 50s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6a92db4569b54847b000f47dfe6b3230 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e89b779a1964facb3bcca8f375412ac06c93626c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-06 12:08:00.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"648d8508432c9ec68691ed2760799c8a9d002dcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 12:43:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65b66cf3460749c7aaf7190c4d55f345\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c02a2009b6334eee9144c1d915595702 : SUCCESS in 6m 04s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/a55dd4237fa04c3c80b0c22c6ba5b39d : SUCCESS in 5m 12s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/12f70ae8d63644df86d8b6850fe05216 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac613dfe52e841d586855f1c57e9e15b : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02b933e726f847879d849cffb33dd148 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0c5394d00c44fab91e5a5303a41c9a0 : SUCCESS in 8m 05s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/9d2f7132bb104a7a941af723acdbf580 : SUCCESS in 29m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/48137b2f914f492b84cce9d5372a39b5 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":2},{"id":"9d4b20c7e9f6ca250058fe60df45601406f6fe2a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-06 12:44:22.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"43a4ed99215b2eb1c4ecac02ef7c3df8ce12e1e6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-02 08:27:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"68422391df70b7f3fcaf88a121588390c0cc6a58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 08:28:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"332ce2923d81a7e939837e07c29428f9d91c0aca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 08:40:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be378fa620394753b11f74935767cb42\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8e86554f3ed94d269d04468c449dcf39 : SUCCESS in 3m 33s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/e567e05bdee64a09a7d09418338eb3cd : SUCCESS in 4m 45s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/9835ff576ef34fa18f9deb3d410c3d05 : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0d855a32d464d23a2bb71ca12cb81de : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f3e19b1109f84aa4ba6ffd3ba36f5a23 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33fe17fd053845e5840157ba4797ff15 : SUCCESS in 7m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/33f0ce5bc136453ab45d945176075c08 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"49a0253cff8808d2644959148d2eea6590baea95","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 08:40:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9cee9c70754c5d12ee7947470b4b3cd859af75da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 08:41:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2a0706204394574ac2590665e6c724c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8436ee84b437482582c02cb1fdb6a96a : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a04df639e518cc3c8214345d37f4073efcf96121","revisions":{"b2050b4998a0310f1ff2bf7104f825cbb5ecdc89":{"kind":"REWORK","_number":1,"created":"2026-02-06 11:19:34.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/93/975893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/93/975893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/93/975893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/93/975893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/93/975893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/93/975893/1"}}},"commit":{"parents":[{"commit":"0fb100ea08ea66b43a6176c67cf247e20bb3a8ae","subject":"Add help text for session start and end date fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/0fb100ea08ea66b43a6176c67cf247e20bb3a8ae"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-06 11:19:34.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-06 11:19:34.000000000","tz":0},"subject":"Add interval unit/value selector to session creation","message":"Add interval unit/value selector to session creation\n\nAlign session scheduling with job UI by adding interval\nunit dropdown, value field.\n\nChange-Id: I57156d0f3fdb0702431d14b35f58266b68853d8b\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/b2050b4998a0310f1ff2bf7104f825cbb5ecdc89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/b2050b4998a0310f1ff2bf7104f825cbb5ecdc89"}]},"branch":"refs/heads/master"},"a04df639e518cc3c8214345d37f4073efcf96121":{"kind":"REWORK","_number":2,"created":"2026-02-06 12:08:00.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/93/975893/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/93/975893/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/93/975893/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/93/975893/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/93/975893/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/93/975893/2"}}},"commit":{"parents":[{"commit":"0fb100ea08ea66b43a6176c67cf247e20bb3a8ae","subject":"Add help text for session start and end date fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/0fb100ea08ea66b43a6176c67cf247e20bb3a8ae"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-06 11:19:34.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-06 12:06:18.000000000","tz":0},"subject":"Add interval unit/value selector to session creation","message":"Add interval unit/value selector to session creation\n\nAlign session scheduling with job UI by adding interval\nunit dropdown, value field.\n\nChange-Id: I57156d0f3fdb0702431d14b35f58266b68853d8b\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/a04df639e518cc3c8214345d37f4073efcf96121"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/a04df639e518cc3c8214345d37f4073efcf96121"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
